)]}'
{"id":"openstack%2Fsushy-tools~720807","triplet_id":"openstack%2Fsushy-tools~master~I6e7e3792f2cede28ea9098ba546f415cec0ddf9b","project":"openstack/sushy-tools","branch":"master","topic":"lower-constraints","hashtags":[],"change_id":"I6e7e3792f2cede28ea9098ba546f415cec0ddf9b","subject":"Stop configuring install_command in tox.","status":"MERGED","created":"2020-04-17 18:46:23.000000000","updated":"2020-04-29 11:22:56.000000000","submitted":"2020-04-29 11:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720807-1588159176099-71cafbf4","meta_rev_id":"76d758ecbd72abfd49fa4d1c486eebef853631bb","_number":720807,"virtual_id_number":720807,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 11:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-22 14:59:27.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-04-27 16:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"date":"2020-04-21 10:11:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2020-04-29 10:12:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 10:11:41.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-04-29 10:12:10.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-29 11:19:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5bd81b79e593ae75b22cac4d2cc750a2dd45aec","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-17 18:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f874debf7799c25c18f20167bd1afcddcfd6fc7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 19:44:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d669d3298ca4412fa9178d57b7a31971 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b7f837247cd46a9a611a69aad11a64d : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a76b93341754fc6a0070860cf0d416f : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecdbcf57737943679de0888b61e94e07 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/102b38e5580b4fa68d97d32358af4869 : FAILURE in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562a954249024c89a406cea5bc95a3f2 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a38acd1c6d34683b41fb033a6fe4639 : SUCCESS in 5m 12s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/46b8b1474fa446f4896075953bab29eb : FAILURE in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"853b63e463edd3980049ef0a576d11dd4ba25711","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-20 16:12:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cd77762747a37575e081438c35e7f3a79a77916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 17:11:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a5e543aa7f74927a5b0f9767f746175 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96a2d90e00944daf8ad6bed23d40857c : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46fdcd98b075486fb9b22b41dd6543b6 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17b279b66c8a4f65b36670639e2e7860 : FAILURE in 5m 17s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/a247ca1bbf3b454da3aa0e83d80b771b : FAILURE in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f395b1d10dd42f9afd489948fecc95a : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc48a29691b54157bf7bd50b3d384256 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ed09faac4742d78a101c058c7b18a2 : SUCCESS in 6m 45s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/7502898d69594142bc6ff3f03352ba1b : FAILURE in 53m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fbc49177015b2469b1090ef9568dce6d92cf38c9","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-21 07:46:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c49a5ae417729ad49ba818f35b7bdefefa8f0e5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:34:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a244d795896c458faa69edddfb559002 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cd3f09faadf4733882d78955a5e0334 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2a80ad5150249eb92ea3fbc768c862c : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86c7f3bbe5df4ecbb57142de677a8977 : FAILURE in 4m 48s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a5af5d873ff48d983db85495545ac9a : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d2c2e86eeb4c78ae8b1557ce45ba13 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b654e0571284e7884827683ba49be72 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6d005a621c4a629ddffd890a03237b : SUCCESS in 6m 01s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/de39fb46fb3a42eeabd58e23bf051aa7 : FAILURE in 45m 20s","accounts_in_message":[],"_revision_number":3},{"id":"1519dbe8a9809a2b5bd0c08a6375c2905ce53950","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-22 14:59:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e19db1b7fd8bb4480cf67ae00a1826340188a3d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:03:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22be4b9b812849e3969c612e9fbe726f : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4a34269a12d43808371285f25bb0efb : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/386aa78f89c84a0ca8ef7cc11937c883 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83982e17882043939428f221b5cd4942 : FAILURE in 5m 06s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/8281db63695f439dad17f35ced4a90b5 : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b78e0748ae44ea5bf5974c47c6201da : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c95e9b049bdd4db593020dd4392d024e : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7291c0cc8a14b29aba1f68652673cf3 : SUCCESS in 6m 52s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d5e9a27d495a4a5ab8c4a516d7939b46 : FAILURE in 51m 36s","accounts_in_message":[],"_revision_number":3},{"id":"c2d1c4928761ff981a06541e33097991a863d6a3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-27 16:04:40.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"652a87607e14c87f7342765692682802b1293f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:06:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c55c70edee4bff88e16218ce78142a : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22a7ff8f3c0e4aadbc19e7476204b023 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d35eadc02924b498f34daebc00ec30f : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5e85980940042dc8f74eab94ebe8bbc : FAILURE in 8m 18s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/78e1343dd5494a6ebca4a6871a8f7dfd : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/266c45f3c3d94825bd006a2dbed1e910 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8c3d08cf05f4efeadc2ec67687d35da : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd6818e81b5423c823346e646e920e8 : SUCCESS in 8m 39s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d35f08cec01a4a03b005e50b3b914b6f : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c4a1fa49628bab9a1ff4b76cd08ae0ba0a94407f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-29 10:12:10.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"33683b4269950fcda23df9569b93c9c2df3b2c57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:12:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e5c68c9c2016779b600583e816744de86648ded","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:19:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e3e0954bfe4f93b7dd2167f7f20a91 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7321dffc3633429388d30e48e5e3bfb4 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4ca2789f40846cabc8e40b24544f981 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3347232a804f4bb085ba9cadb2f77c19 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22dcdd7c6cb44381a5735a35dd32e113 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d776d19b2444f3a909cbd1536137230 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bd9b08d6634349bed58abf13f85dd5 : SUCCESS in 5m 41s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/a78fcccbd5a0491a98af52c7fb967c7b : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9f6ba885e11ca5a0db3fa6ddff63569691d1baed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:19:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"58f8ef4f2513af7211325f947350265905c1096f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:22:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91eb6c825ff949b18685723aa6c9a1fa : SUCCESS in 1m 14s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f57ef996684be48e775dd5f6c4c042 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"73ad654b013e648280671bcb14259b63d709677f","revisions":{"f66c9d79efa54cac32f298c1c5acf11543963ef8":{"kind":"REWORK","_number":1,"created":"2020-04-17 18:46:23.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/07/720807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/07/720807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/1"}}},"commit":{"parents":[{"commit":"ab975120028e3064f6d175aabbb9a33e1a6aa518","subject":"Convert sushy-tools tempest to dib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ab975120028e3064f6d175aabbb9a33e1a6aa518"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-17 18:46:04.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-17 18:46:04.000000000","tz":120},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nopenstack-discuss thread\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014237.html\n\nChange-Id: I6e7e3792f2cede28ea9098ba546f415cec0ddf9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f66c9d79efa54cac32f298c1c5acf11543963ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f66c9d79efa54cac32f298c1c5acf11543963ef8"}]},"branch":"refs/heads/master"},"c9c7de1790b7bafdee612836387f2eef1755e984":{"kind":"REWORK","_number":2,"created":"2020-04-20 16:12:00.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/07/720807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/07/720807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/2"}}},"commit":{"parents":[{"commit":"ab975120028e3064f6d175aabbb9a33e1a6aa518","subject":"Convert sushy-tools tempest to dib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ab975120028e3064f6d175aabbb9a33e1a6aa518"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-17 18:46:04.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-20 16:11:19.000000000","tz":120},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nopenstack-discuss thread\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014237.html\n\nBumping l-c:\n-openstacksdk to 0.13.0\n-Werkzeug to 0.14\n\nChange-Id: I6e7e3792f2cede28ea9098ba546f415cec0ddf9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c9c7de1790b7bafdee612836387f2eef1755e984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c9c7de1790b7bafdee612836387f2eef1755e984"}]},"branch":"refs/heads/master"},"73ad654b013e648280671bcb14259b63d709677f":{"kind":"REWORK","_number":3,"created":"2020-04-21 07:46:56.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/07/720807/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/07/720807/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/07/720807/3"}}},"commit":{"parents":[{"commit":"ab975120028e3064f6d175aabbb9a33e1a6aa518","subject":"Convert sushy-tools tempest to dib","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ab975120028e3064f6d175aabbb9a33e1a6aa518"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-17 18:46:04.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-21 07:46:25.000000000","tz":120},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nopenstack-discuss thread\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014237.html\n\nBumping l-c:\n-openstacksdk to 0.13.0\n-Werkzeug to 0.14\n\nBumping t-r:\n-openstacksdk to 0.13.0\n\nChange-Id: I6e7e3792f2cede28ea9098ba546f415cec0ddf9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/73ad654b013e648280671bcb14259b63d709677f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/73ad654b013e648280671bcb14259b63d709677f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
