)]}'
{"id":"openstack%2Fsushy-tools~728048","triplet_id":"openstack%2Fsushy-tools~master~Ifed01ea2a379672875d0a5d5e33096bda9bd36ed","project":"openstack/sushy-tools","branch":"master","topic":"fix_pdfbuild","hashtags":[],"change_id":"Ifed01ea2a379672875d0a5d5e33096bda9bd36ed","subject":"Fix pdf build","status":"MERGED","created":"2020-05-14 11:43:43.000000000","updated":"2020-05-19 18:34:32.000000000","submitted":"2020-05-15 09:06:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728048-1589533586865-be6e565f","meta_rev_id":"bc3cad45a58e49924c2270cefe745128213ca533","_number":728048,"virtual_id_number":728048,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 09:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-05-15 08:01:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-14 16:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-05-15 08:01:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 16:45:11.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-15 08:01:12.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-05-15 09:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14d5f47241740da43843281ec15cdbb1b7620c79","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-05-14 11:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9db6905a064b5c9c4dffffeafc3e8523098fc245","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-14 12:07:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93e5b8e02845d7a65a480670a69ea27d119641a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 12:42:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e3c8dcc64f4e49bc60bb47718af89b : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac2f3b19658140c8b2820308a11fbc42 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afb563c3f78240caa45f8202f99ee591 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cde7a3c4d0044ed3a2b6b94eff875d51 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1556f714ca3f4f7d849508ed559db41f : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6f201e84d214083b8dcd65f1cef6b90 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6923fec06d934f75a4595a79094901c3 : SUCCESS in 7m 07s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d6b92d9883514da69f2830258d8db41f : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b1304da2257414251ff99e82c135a73a0e49087f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-05-14 14:16:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a98a431272d99230962a1568c0c0c35348fd439b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 15:25:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56af4d20e89a4409a53d9590df6e0d26 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe9826df320c4e6d8fd7a29ee2ef3871 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6e8b6aa4e774982bb721b148ed4b9dd : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16650bb216704695bac386f4a2c006bf : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93a3491b1ec64cd9ba683b514b45171a : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73e06733996c4485b54bb010eba60264 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8712ab27d9c4fdc999137b4fe77d782 : SUCCESS in 10m 18s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d33be314b6674f52854bee984d592882 : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":2},{"id":"327842ca924e083b73c1ebc9f8c8a97eab148345","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-14 16:45:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"033013da48894df1dd35525ffd019888574a13c2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-15 08:01:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ffc7696c2168b43a516b04b47b82c413ca8d3fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 08:01:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5dcfc440a1be443e6f3b2df13078f34f92a1630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 09:06:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdc242646f34633a44587c0e19272ac : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0757e1a77ef94ac1941b2abc5c8331c6 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97455025e4a440d5833e7ebc5e57df5d : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45ecc19198494bbe814f6087dce82f43 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6ba17c1fa3544ac9a947ddeadf45979 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/813cbeaf36b749dab028938f73155c41 : SUCCESS in 8m 02s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/c94ac15d8025416c8313f2c805101c98 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a9035932970fff4cb399ee4dc8e95db5a6c95dbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 09:06:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4d795f314508b05428bf8a9fdc2d47a85fbf35e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 09:08:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c00f7cfb0da4160bd13546c9e55028a : SUCCESS in 1m 33s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddac487c14514717beaf37c055fa8908 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2},{"id":"78b2a4ccd36ff65c76ebd039b0dbd06cd40d7d2f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-19 18:34:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d72c2e95444fb3045b53764f40ec4b4fad3975e","revisions":{"199054fb0b2477a989fb5d5452047d5d87397a6c":{"kind":"REWORK","_number":1,"created":"2020-05-14 11:43:43.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/48/728048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/48/728048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/1"}}},"commit":{"parents":[{"commit":"e70cbf8d233e5229e1b0f36e496d10e2e98b1b69","subject":"Update lower-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e70cbf8d233e5229e1b0f36e496d10e2e98b1b69"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-05-14 11:43:25.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-05-14 11:43:25.000000000","tz":120},"subject":"Fix pdf build","message":"Fix pdf build\n\nopenstackdocstheme 2.1.0 broke the pdf builds, adding a workaround\nthat fix the problem and also works with older openstackdocstheme.\n\nChange-Id: Ifed01ea2a379672875d0a5d5e33096bda9bd36ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/199054fb0b2477a989fb5d5452047d5d87397a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/199054fb0b2477a989fb5d5452047d5d87397a6c"}]},"branch":"refs/heads/master"},"7d72c2e95444fb3045b53764f40ec4b4fad3975e":{"kind":"REWORK","_number":2,"created":"2020-05-14 14:16:47.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/48/728048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/48/728048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/48/728048/2"}}},"commit":{"parents":[{"commit":"e70cbf8d233e5229e1b0f36e496d10e2e98b1b69","subject":"Update lower-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e70cbf8d233e5229e1b0f36e496d10e2e98b1b69"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-05-14 11:43:25.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-05-14 14:16:35.000000000","tz":120},"subject":"Fix pdf build","message":"Fix pdf build\n\nopenstackdocstheme 2.1.0 broke the pdf builds, adding a workaround\nthat fix the problem and also works with older openstackdocstheme.\n\nChange-Id: Ifed01ea2a379672875d0a5d5e33096bda9bd36ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7d72c2e95444fb3045b53764f40ec4b4fad3975e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7d72c2e95444fb3045b53764f40ec4b4fad3975e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
