)]}'
{"id":"openstack%2Fsushy-tools~733152","triplet_id":"openstack%2Fsushy-tools~master~Ifcc9afc46562a5df5dd73010b8d274452c1aefc1","project":"openstack/sushy-tools","branch":"master","hashtags":[],"change_id":"Ifcc9afc46562a5df5dd73010b8d274452c1aefc1","subject":"Read current domain state from XML on filesystem","status":"MERGED","created":"2020-06-03 13:55:43.000000000","updated":"2020-06-05 14:20:15.000000000","submitted":"2020-06-05 14:17:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733152-1591366648555-f48fbef3","meta_rev_id":"0786c0a269ed77d7a0f0ef7d02c27855fe03f204","_number":733152,"virtual_id_number":733152,"owner":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 14:17:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-06-05 13:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 08:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2020-06-05 13:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 08:56:35.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-06-05 13:19:35.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-06-05 14:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"963567aa70a042979299fe07f30e74c08076f4c8","author":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"date":"2020-06-03 13:55:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3b85b6a92a98f45aed046be50a2d39af9d012b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 15:12:01.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed8dc62f7c644186b798eaa3fa8d806f : FAILURE in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3638925388de4b0b82afc41a0cdb0bdf : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bb2444955da43e1858e75e656c3780b : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/935556b7be2a4aeb83b76fe8a2e9e102 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9a293c1e7d0462da6fee2219eb35d26 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ca1be4f73c34208aa169c4e6756225a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5798c15e660246a2978f06bb7c9d045e : SUCCESS in 6m 30s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/4753df3d80c641028328f8e780bbe08a : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"344a6b2d46c03a6897fa9037e67d332572b53f6f","author":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"date":"2020-06-03 17:57:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2d68316a366f08f97eb18ad34781d3661e54265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 19:05:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e1986fe1f34ae1ba5f39365688daa5 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bed49264124432c9bf1fcdf119e7a99 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0559f283e62844d1beec9e7be6231b9c : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c769ba0edcd48aaadc5c538702f6035 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9ead4b0fed34c75855c3aa690b6665c : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d61d259ecdbd4ea68b87e94417a484b4 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5cde38eded4f98a2966138d53f305a : SUCCESS in 13m 09s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/21d9cf12f34349b48eace2bd1213b948 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5effc4f37f955b57ec341f1b493d525a72a4e69a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-04 08:56:35.000000000","message":"Patch Set 2: Code-Review+2\n\nNice!","accounts_in_message":[],"_revision_number":2},{"id":"ca59a3c146e6da3b52dd0a08694c9503665ed95a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-06-05 13:19:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97d8261f49da0b2bb0a7bfe8e8b059ec798ba4b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:19:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c93c83a3a7d0aa0d3797d5564296323f98a43b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:17:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac18c1cd7bf74329874b885c0cc18580 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97488aee8dae47678a75d8e7f5467669 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de16eb9bf883432ba2850ba1c4facc51 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3672b8d240d4f9a82067f0e3a25b567 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1012b3b2d4a94d87b94f3b91c43a5595 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66fee862de04fa5a1c3987ba1002810 : SUCCESS in 7m 20s\n- sushy-tools-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/458a32e8ab8e4f77aea7b762169fed39 : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5c47c50a8e0a4469ddea18bb9b07a985640e9faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:17:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0786c0a269ed77d7a0f0ef7d02c27855fe03f204","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:20:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6683c596f75421880225d5439345aa6 : SUCCESS in 2m 24s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19365dbae7ec47db880091327ede74c1 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13a0b5920c13d86ef1dad89df522be4606993a39","revisions":{"ac8425358d7c6a25707d2b2d104dfa295956d1da":{"kind":"REWORK","_number":1,"created":"2020-06-03 13:55:43.000000000","uploader":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"ref":"refs/changes/52/733152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/52/733152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/1"}}},"commit":{"parents":[{"commit":"6baebba0324176b3043cb82519f935c202a3ce1e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6baebba0324176b3043cb82519f935c202a3ce1e"}]}],"author":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2020-06-03 13:33:41.000000000","tz":60},"committer":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2020-06-03 13:55:35.000000000","tz":60},"subject":"Read current domain state from XML on filesystem","message":"Read current domain state from XML on filesystem\n\nUse the XML file as a source of truth when getting current\nstate of each domain. Reading current state from the running\nVM\u0027s requires that we do a reboot every time we change something.\n\nChange-Id: Ifcc9afc46562a5df5dd73010b8d274452c1aefc1\nStory: #2007735\nTask: #39901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ac8425358d7c6a25707d2b2d104dfa295956d1da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ac8425358d7c6a25707d2b2d104dfa295956d1da"}]},"branch":"refs/heads/master"},"13a0b5920c13d86ef1dad89df522be4606993a39":{"kind":"REWORK","_number":2,"created":"2020-06-03 17:57:01.000000000","uploader":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"ref":"refs/changes/52/733152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/52/733152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/52/733152/2"}}},"commit":{"parents":[{"commit":"6baebba0324176b3043cb82519f935c202a3ce1e","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6baebba0324176b3043cb82519f935c202a3ce1e"}]}],"author":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2020-06-03 13:33:41.000000000","tz":60},"committer":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2020-06-03 17:56:52.000000000","tz":60},"subject":"Read current domain state from XML on filesystem","message":"Read current domain state from XML on filesystem\n\nUse the XML file as a source of truth when getting current\nstate of each domain. Reading current state from the running\nVM\u0027s requires that we do a reboot every time we change something.\n\nChange-Id: Ifcc9afc46562a5df5dd73010b8d274452c1aefc1\nStory: #2007735\nTask: #39901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/13a0b5920c13d86ef1dad89df522be4606993a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/13a0b5920c13d86ef1dad89df522be4606993a39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
