)]}'
{"id":"openstack%2Fsushy-tools~738874","triplet_id":"openstack%2Fsushy-tools~master~Ia926a8f9d8debaff9763aeebcd986f9056cca700","project":"openstack/sushy-tools","branch":"master","topic":"738874","hashtags":[],"change_id":"Ia926a8f9d8debaff9763aeebcd986f9056cca700","subject":"Support non-ide virtual media buses","status":"MERGED","created":"2020-07-01 16:01:29.000000000","updated":"2020-07-02 22:02:46.000000000","submitted":"2020-07-02 22:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738874-1593727203701-8c5a5b72","meta_rev_id":"583d352e35ed933f677c035bb12013c83581f9e6","_number":738874,"virtual_id_number":738874,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-02 16:52:26.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 22:00:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-07-02 17:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},{"value":2,"date":"2020-07-02 20:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},{"value":1,"date":"2020-07-02 20:45:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 20:54:33.000000000","updated_by":{"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},"reviewer":{"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},"state":"REVIEWER"},{"updated":"2020-07-02 13:48:11.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-07-02 17:10:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-07-02 20:45:02.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-07-02 22:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85c30cf5f6f5464bec17459d35d0754d1079d6b7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-01 16:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2513c34c8f1628f3bfdd404b49ba349e8d80e7cf","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-01 16:35:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bafd3d8a025c9aa569c789c3e0ad114e0b506694","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-01 17:13:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e120f0b030a1acd5311c8f1abdd68d30cb3f91ad","author":{"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},"date":"2020-07-01 17:57:22.000000000","message":"Patch Set 3: Code-Review+1\n\nThis worked for me; my Q35 libvirt machine got a SCSI cd-rom drive added with the correct unit number. Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"08bc87c15cfe292122afc6d7b83852fee610b15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 18:09:09.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f75ebe5d4977421e8b5790e176c78e84 : FAILURE in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5136df3983b4cbcb8b77b26406042d2 : SUCCESS in 8m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6b7b042d73244cf89a109364a682b51 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/969ccd58941244ae984c3d7cd9c8c751 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/490d49eb99ca4e6d9d3be60a39db760c : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f562202823f64a8d96567da1d282bdaa : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e22dfefd8d34672aaa0b5b97b8c6242 : POST_FAILURE in 36m 27s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/160e9dd5daff43b9a81ab622e3ffb1a5 : SUCCESS in 50m 00s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d1f7462daa314175b07061b97dc8a264 : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":3},{"id":"67d01c34bdf4cbf9564b1eee14fd28ff1e8d6f0b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-01 18:55:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"82c763c5d68b331c150a19112097da3c9f67057e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 20:16:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f0db73668f4b919cbaf9cfcaf18a2a : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2071ad11bc9344838e8349ac88a9c9b7 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a86d0afa0af4581a6256b93bdd67654 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a6f5f7d01f410c9205085f7bf7f98e : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bf3d3b138f242c187cacd72ed357042 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1694f4baa88474e8ddc8966f9892184 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c7747078f74d0abb5ac43c238e892e : SUCCESS in 5m 29s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/32a9036712d6451abda33758f88a892f : SUCCESS in 51m 24s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/29375b69d31a4c4d96e9f354aa5ee604 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":4},{"id":"e8714e6bcb89f65b92bdea3ce8d45d4827465340","author":{"_account_id":32157,"name":"Stephen Benjamin","email":"stephen@redhat.com"},"date":"2020-07-01 20:54:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"33e969a1303247a53f291d1a5107b8c6de902fea","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-01 21:18:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0afce4060ea08cbeddd9637b4e10f20cab240add","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-02 08:09:18.000000000","message":"Patch Set 4: Code-Review-1\n\nPlease add unit tests.","accounts_in_message":[],"_revision_number":4},{"id":"4c281b23698bee3b0e1c9a43ade2fab4b6365cc2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-02 13:01:52.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"893519ebafd8232dc39fe0f8964d7f1d21073b7b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-02 13:48:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a007dc153497862d682387463eb11d3fa9677500","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-02 13:48:11.000000000","message":"Patch Set 4: -Code-Review\n\nRemoving my -1 based on IRC conversation where we found out that the current unit test coverage is too spotty to meaningfully insert a new one.","accounts_in_message":[],"_revision_number":4},{"id":"3c1ac5ea293146248f154cda75645b30179bd74a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-02 16:18:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"53ff191a035968fa0baee24f6d0a753c82424175","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-02 16:25:06.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9a3840aa4fa9bf570971b14c37d4d8cf6f88e3da","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-02 16:52:26.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2f6154e9ecec00aab50b31f380c678de2eebb5e4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-02 17:10:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b7def13001253a980b5d29c20526883902ae98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:17:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9af9ecfcdf640ddb11018a65e09b36a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efc08730b0014171b9d36c261c9e8613 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ce6c1b7e0d647f7870ac66efa989ef6 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae068d017cef4ac693945e836521dc42 : FAILURE in 4m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a05a37d5d52f4e698bf0d311fede0bf1 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0d79d9923814153a9ef7c3f91a5d377 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48064d32fee4d9cbc44139758468bb3 : SUCCESS in 9m 13s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0f7c5d2eac8b46ed9d80edf2db321325 : SUCCESS in 34m 11s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f7a0284db6064b04b4b468afc74e65ca : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":5},{"id":"d38e4e5742f8171e9a4850ab988fa94d64911e2f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-02 20:45:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nApproving so we can move forward, I will work on a fix for py38 on the tests.","accounts_in_message":[],"_revision_number":5},{"id":"bc05125aadc8ed9745953c2bf28ec4ce0dafc08b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:45:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f3f44e05d4fc9c42195afd723cc3811c79239e24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 22:00:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"30ad03012465c3fed36e98ffe87814322a4725ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 22:00:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ab432ca4e4a4fd8a5930ffaf2c96d6f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e44c699cd66647cab218716ffbeee729 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1101410438f4148bf6c5e9037d3fcf5 : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f67f36dfc40d4127a21a986c025bc024 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b139fd243e11497a9964d3815396e492 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904586614537481ab100c25acaeee306 : SUCCESS in 6m 15s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3a182f87e13b49219be0b5cc09f20a51 : SUCCESS in 1h 07m 51s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/698ab42e2cec49849c1d87bccb60dfa6 : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":5},{"id":"27adc85ac5123a6cb7ef1a63f58ad52a49aa004e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 22:02:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2941e9c070d8461e993917ed7737fb54 : SUCCESS in 2m 22s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e841681150945dd8143e922c1540ced : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ff2be838b9da25998202a93f2120d1350f2a1f6e","revisions":{"3460f32272edefababbddfb9a4c84dde9d1323e5":{"kind":"REWORK","_number":1,"created":"2020-07-01 16:01:29.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/74/738874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/738874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/1"}}},"commit":{"parents":[{"commit":"74769fac826f486742a4ee2e0e177b5a72677c49","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/74769fac826f486742a4ee2e0e177b5a72677c49"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:01:12.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:01:12.000000000","tz":-420},"subject":"WIP: support non-ide virtual media buses","message":"WIP: support non-ide virtual media buses\n\nChange-Id: Ia926a8f9d8debaff9763aeebcd986f9056cca700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3460f32272edefababbddfb9a4c84dde9d1323e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/3460f32272edefababbddfb9a4c84dde9d1323e5"}]},"branch":"refs/heads/master"},"61463ae9f8c1c76e063609bbbe3ae58e097bdded":{"kind":"REWORK","_number":2,"created":"2020-07-01 16:35:11.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/74/738874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/738874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/2"}}},"commit":{"parents":[{"commit":"74769fac826f486742a4ee2e0e177b5a72677c49","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/74769fac826f486742a4ee2e0e177b5a72677c49"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:01:12.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:35:06.000000000","tz":-420},"subject":"WIP: support non-ide virtual media buses","message":"WIP: support non-ide virtual media buses\n\nChange-Id: Ia926a8f9d8debaff9763aeebcd986f9056cca700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/61463ae9f8c1c76e063609bbbe3ae58e097bdded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/61463ae9f8c1c76e063609bbbe3ae58e097bdded"}]},"branch":"refs/heads/master"},"e77dccdf37e4b2b152bc2cf9bb668dcba5cfb08e":{"kind":"REWORK","_number":3,"created":"2020-07-01 17:13:20.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/74/738874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/738874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/3"}}},"commit":{"parents":[{"commit":"74769fac826f486742a4ee2e0e177b5a72677c49","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/74769fac826f486742a4ee2e0e177b5a72677c49"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:01:12.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 17:13:14.000000000","tz":-420},"subject":"Support non-ide virtual media buses","message":"Support non-ide virtual media buses\n\nStory: 2007881\nTask: 40249\n\nChange-Id: Ia926a8f9d8debaff9763aeebcd986f9056cca700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e77dccdf37e4b2b152bc2cf9bb668dcba5cfb08e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e77dccdf37e4b2b152bc2cf9bb668dcba5cfb08e"}]},"branch":"refs/heads/master"},"2b81e4021a6213f44425697912c6a65618438b0b":{"kind":"REWORK","_number":4,"created":"2020-07-01 18:55:43.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/74/738874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/738874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/4"}}},"commit":{"parents":[{"commit":"74769fac826f486742a4ee2e0e177b5a72677c49","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/74769fac826f486742a4ee2e0e177b5a72677c49"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:01:12.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 18:55:38.000000000","tz":-420},"subject":"Support non-ide virtual media buses","message":"Support non-ide virtual media buses\n\nStory: 2007881\nTask: 40249\n\nChange-Id: Ia926a8f9d8debaff9763aeebcd986f9056cca700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2b81e4021a6213f44425697912c6a65618438b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2b81e4021a6213f44425697912c6a65618438b0b"}]},"branch":"refs/heads/master"},"ff2be838b9da25998202a93f2120d1350f2a1f6e":{"kind":"REWORK","_number":5,"created":"2020-07-02 16:18:53.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/74/738874/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/74/738874/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/74/738874/5"}}},"commit":{"parents":[{"commit":"74769fac826f486742a4ee2e0e177b5a72677c49","subject":"drop mock from lower-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/74769fac826f486742a4ee2e0e177b5a72677c49"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-01 16:01:12.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-02 16:18:52.000000000","tz":-420},"subject":"Support non-ide virtual media buses","message":"Support non-ide virtual media buses\n\nAlso fix a minor possible bug with interpretation of XML returning\na string trying to be compared to a number, and adds two test files\nand associated tests for the injection of SATA or SCSI removable\ndisk drives.\n\nStory: 2007881\nTask: 40249\n\nChange-Id: Ia926a8f9d8debaff9763aeebcd986f9056cca700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ff2be838b9da25998202a93f2120d1350f2a1f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ff2be838b9da25998202a93f2120d1350f2a1f6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
