)]}'
{"id":"openstack%2Fsushy-tools~744471","triplet_id":"openstack%2Fsushy-tools~master~Ie81bef0eca87d8bcd2d8302df3186d79fb7598f3","project":"openstack/sushy-tools","branch":"master","hashtags":[],"change_id":"Ie81bef0eca87d8bcd2d8302df3186d79fb7598f3","subject":"Fix handling SUSHY_EMULATOR_CONFIG when running under a WSGI server","status":"MERGED","created":"2020-08-03 13:38:27.000000000","updated":"2020-08-03 16:52:11.000000000","submitted":"2020-08-03 16:50:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744471-1596473450884-08ff93a1","meta_rev_id":"b0d835faddc1b1fd899e57d213a10fb4c66d0636","_number":744471,"virtual_id_number":744471,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 16:50:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-08-03 15:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-08-03 15:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-08-03 15:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 15:01:09.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-08-03 15:24:20.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-03 16:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aadcf9bd924dca23c81a359c8e49e33e73d2e752","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-03 13:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c71c63ebf87b614556fdeadd992bdd7ec2e183a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:57:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddcf565c53a645e89480a7e188de1c6a : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d61b3b9f465b48aba5717e60964fb03a : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee4efac532724e5a8edd0fbe89041507 : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2da76f5b2a3a44d9bc713e7893872196 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aac6c9f0c274267aed985e014f68ad2 : SUCCESS in 5m 20s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/47e51b09815c42c4bb3eee396fc997e8 : SUCCESS in 36m 24s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/84fa553e7a134de78b2f1ae26fda811d : SUCCESS in 1h 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5e7429e93d51b613ad7df5320257608b97524782","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-03 15:01:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a99d9329665d5d91bdf6e9de0ef0c7a46fbd0a00","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-03 15:24:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8a8616566925a8928647586fb4cbe9187dbdb9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:24:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6758fda56f5cbee901148b69654767254318cc27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:50:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d459021b5c6f3d92ca1b55d630458040d1d71ad5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:50:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be51b7f877a34421ac25df544e11ae8a : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfbd9f55ff4c439b974eeac61ba09f92 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9447de966e93497da62518e4df5a9c7b : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e5191deee544427813a922fe660204c : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4d1ce8ccdd4eaebfe68068b32c61d3 : SUCCESS in 4m 40s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3a6cdf2df2a24c51aec81083d5479b5f : SUCCESS in 1h 13m 42s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ec33814e27614bed95362f31ce92c46d : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b0d835faddc1b1fd899e57d213a10fb4c66d0636","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:52:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54f7667c1f148059e01dfb5c44f7283 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4054cf8e91ae8fd8c7d42ab8581dc17fa31ddbd","revisions":{"f4054cf8e91ae8fd8c7d42ab8581dc17fa31ddbd":{"kind":"REWORK","_number":1,"created":"2020-08-03 13:38:27.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/71/744471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/71/744471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/71/744471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/71/744471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/71/744471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/71/744471/1"}}},"commit":{"parents":[{"commit":"7edb57ddc177f8e5416fd5fbbaae358242987913","subject":"Merge \"Support non-ide virtual media buses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7edb57ddc177f8e5416fd5fbbaae358242987913"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-03 13:34:50.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-03 13:34:50.000000000","tz":120},"subject":"Fix handling SUSHY_EMULATOR_CONFIG when running under a WSGI server","message":"Fix handling SUSHY_EMULATOR_CONFIG when running under a WSGI server\n\nThis variable is currently handled in main() which is not run when\na real WSGI server is used, making it impossible to pass any config.\n\nChange-Id: Ie81bef0eca87d8bcd2d8302df3186d79fb7598f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f4054cf8e91ae8fd8c7d42ab8581dc17fa31ddbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f4054cf8e91ae8fd8c7d42ab8581dc17fa31ddbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
