)]}'
{"id":"openstack%2Fsushy-tools~744862","triplet_id":"openstack%2Fsushy-tools~master~I46dd794750c182572d37bab3ebfcd608ceb1bb6c","project":"openstack/sushy-tools","branch":"master","topic":"fake-manager","hashtags":[],"change_id":"I46dd794750c182572d37bab3ebfcd608ceb1bb6c","subject":"Replace the static Manager driver with a new fake one","status":"MERGED","created":"2020-08-05 09:13:31.000000000","updated":"2020-08-06 21:42:04.000000000","submitted":"2020-08-06 21:40:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744862-1596750033385-2e16cef6","meta_rev_id":"35e90a5251ca1b4dc8e8135dffbed42c84cb3974","_number":744862,"virtual_id_number":744862,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 21:40:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":1,"date":"2020-08-06 12:20:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2020-08-06 20:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-08-06 11:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2020-08-06 20:37:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 11:18:51.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-06 12:20:42.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2020-08-06 20:37:03.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-08-06 21:40:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47637b28414af3c3f31e930e34c84604f7296049","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 09:13:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7121960475f867e75e3141992df0b3e6d82d4727","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 09:16:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef479fea94d067b542f76a4629e438d1ae9abef3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 09:33:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f6c264b2d6bc03e62aae6bf77742cb7d39f410c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:36:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f594ddd79ba04c8798e7f8bd858addaf : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc94153227143958b2b3b269fafef20 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe329ec52db446e8a6388f17972f907 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/755f70f37eb84603ad1bbf4b0aa94725 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e500fa2782472e9a06c1e6f9b7f29c : SUCCESS in 6m 51s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bbccd07ad6a4494fa505dfaeed2c0a81 : SUCCESS in 54m 58s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8a914a5c918046dba910268d27ec7861 : FAILURE in 51m 23s","accounts_in_message":[],"_revision_number":3},{"id":"c1e56a8de6a4da2185397f3a99a830e7f57d08bc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 10:36:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d3024944df0e6e44bbba80d645f2e2788a2ed46c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 11:36:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dbbebdc7a221b74e7d616801cbbb36133638104f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 11:41:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e9b345750d351461ad4350a6627f4df276ffe67b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 11:43:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8d066f44ee18ddfe449201267cb4b9c7dc0dbe00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 12:27:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad66a4eb7c694498880c0d193326c0d0 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23cd051fdb074e0ebcfb013729d5cde4 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b34b0fe2e80e4937a7a74b8cda327671 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/219002dae9154f4fb20e36a8e18f0cc7 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944c58f4aa904023ab225083991bf39d : SUCCESS in 7m 12s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/387fce18ec81491389b56d56c5c03c43 : FAILURE in 31m 07s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a3ebc9075ad74d6d8ecbb8fc8b0ba116 : FAILURE in 40m 56s","accounts_in_message":[],"_revision_number":7},{"id":"d712f7f72cf9b4336454bd960bf5ea6684b9fb9c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 12:37:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f03105c52ce2dc2f13f325c98e68c71cb95cd69c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 13:14:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d302ea1e38326e3203787b3cb96847823624d9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 14:27:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db0e7f451234c50b77f64e7fd36f462 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b48598a2a44154841d99d7da227d0a : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7672947018f141f0859a7be8b51bd6eb : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/727c70c3d9e041df8d255d59f8605695 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e53ccd2a49b4e9caaf05930a23aaec1 : SUCCESS in 7m 14s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bc774113b5fe4d258427fcc7fb4a8891 : SUCCESS in 52m 42s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1c991dbc9cfd4b0ba3c9d3330d8494dc : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":9},{"id":"1c6b801a5db79e48cb28ab639fa6289b2d478039","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-05 17:49:08.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1be56bf0d17baa05f0ed0412edd295ecdea55979","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-06 07:43:27.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f16b86b484c6be3ae0b37a768c99331ba7d9d455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 08:39:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466a84a647e7452796624be21fab733f : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bb93f133cb84e5b9c7d096f83b725f7 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2980afcefe0d4aac84cb6dcd9f7f45f0 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d9c2cd4d774103944c62f6ef51e9a9 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749acef615e440c58cd27b498439d68b : SUCCESS in 5m 33s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c2f95c3578d748408c0edc0bbda7085a : SUCCESS in 39m 23s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8867530ab6d244f09d3eea49b98a7eaa : FAILURE in 46m 07s","accounts_in_message":[],"_revision_number":10},{"id":"805f9cc5411d94bb5ac3ce8757a97ff0595b8156","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-06 08:42:41.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"08f9ce998102b3796bb8e1bbd4a3e2c0a737c174","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-06 09:42:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a364bc150ec9694fe9d906271e2f1f576cdff76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 10:42:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daff6f22faad40a490ac03db55b167a2 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/090d18537f0d4f6894765614470c40af : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fa567ecee8946fe9a9bf6168033bdd3 : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bd265eaf23341ba9732963a166dcd4e : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334672abdf424565bc2199ed3969c5f0 : SUCCESS in 8m 02s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/522b3ac9848a4a16a840fc99c0caf3a8 : SUCCESS in 36m 49s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5a3c0fb3dd86412f96a1a127c25e05fd : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":12},{"id":"d6bc5d300089191aa5297e4c809f5e5084f76651","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-06 10:50:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e59fb0e8a517466ce4025209b95f6343a374ad4a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-06 11:18:51.000000000","message":"Patch Set 13: Code-Review+2\n\nLGTM let\u0027s see what the CI says :)","accounts_in_message":[],"_revision_number":13},{"id":"82bff93bfe708cb5a07200e9cb2298b800fff674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 11:56:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e204c4d4f5fa4149a9ac7de9e178110f : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d1becf2170a4140bb16bfce5362ecc1 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d70ed063c055459080e8aaaacc92ff45 : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79d524e027ed4b73a0701ad810e78c63 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b77f8c3eae4b5292a6a8d53274e049 : SUCCESS in 7m 00s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e0f424453bba4fd5aed13b44e0fed5a3 : SUCCESS in 49m 41s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b21461c2be4d45aa80c73d165bd7b886 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":13},{"id":"ef352f5cef4ce784e415ea79e17c0fcb12d21354","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-08-06 12:20:42.000000000","message":"Patch Set 13: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":13},{"id":"c3175a265254c06ec2e7b3aed26b7881b5084653","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-08-06 20:37:03.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":13},{"id":"2fee4320753415cb4747cca2f54f06994653a23e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:37:14.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"3cd6290204a7d3ec4cc1df407d3382b744e0f073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:40:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"ec40e07a83cfa0569707f1c5fc0a57163cef7983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:40:33.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3ba9b2147544baa0b759517c5de6f1 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/801c38dcb37946c5aec292f7ae6945f4 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/385792b5883d43b29aa1b236c7972960 : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cc6c60c731d410d82e203d32e1b8876 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3feb064564e547108036e302b5cc7ab1 : SUCCESS in 5m 43s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/16f53e898ea34f88b7271fa8eb6d1e34 : SUCCESS in 37m 55s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/736949c65fc54e1db406fb9667c14be0 : SUCCESS in 56m 24s","accounts_in_message":[],"_revision_number":13},{"id":"a8c257430a9f45b4ae03df27fefd72169ca6e05f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:42:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe54487cd7694f48b914c5b7a9f12bca : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"b3bbb5b108216e39e5d6a98ae34e3ec38b568cf5","revisions":{"4d440259d24e94ef3387e732cb709afe3c03d2f4":{"kind":"REWORK","_number":1,"created":"2020-08-05 09:13:31.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/1"}}},"commit":{"parents":[{"commit":"d3229db6c98575b0e7607225961aefedc957e82c","subject":"Merge \"Update version of hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d3229db6c98575b0e7607225961aefedc957e82c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:12:42.000000000","tz":120},"subject":"Simplify the Manager driver","message":"Simplify the Manager driver\n\nUse a get_manager call instead of uuid/name calls that transform\nidentities. Simplify get_manager to not fetch data it ignores anyway.\nMove the (broken) systems and chassis listing logic to the statis\nmanager driver, so that it can be replaced by something working.\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d440259d24e94ef3387e732cb709afe3c03d2f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d440259d24e94ef3387e732cb709afe3c03d2f4"}]},"branch":"refs/heads/master"},"fcc7fc2f51c26cba413d4471e9274ee4164e3fc0":{"kind":"REWORK","_number":2,"created":"2020-08-05 09:16:36.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/2"}}},"commit":{"parents":[{"commit":"d3229db6c98575b0e7607225961aefedc957e82c","subject":"Merge \"Update version of hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d3229db6c98575b0e7607225961aefedc957e82c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:16:27.000000000","tz":120},"subject":"Simplify the Manager driver","message":"Simplify the Manager driver\n\nUse a get_manager call instead of uuid/name calls that transform\nidentities. Simplify get_manager to not fetch data it ignores anyway.\nMove the (broken) systems and chassis listing logic to the statis\nmanager driver, so that it can be replaced by something working.\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fcc7fc2f51c26cba413d4471e9274ee4164e3fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fcc7fc2f51c26cba413d4471e9274ee4164e3fc0"}]},"branch":"refs/heads/master"},"4ac6f453f097afcdfc66e04aae1b733a9f6e348e":{"kind":"REWORK","_number":3,"created":"2020-08-05 09:33:22.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/3"}}},"commit":{"parents":[{"commit":"d3229db6c98575b0e7607225961aefedc957e82c","subject":"Merge \"Update version of hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d3229db6c98575b0e7607225961aefedc957e82c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:32:53.000000000","tz":120},"subject":"Simplify the Manager driver","message":"Simplify the Manager driver\n\nUse a get_manager call instead of uuid/name calls that transform\nidentities. Simplify get_manager to not fetch data it ignores anyway.\nMove the (broken) systems and chassis listing logic to the statis\nmanager driver, so that it can be replaced by something working.\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4ac6f453f097afcdfc66e04aae1b733a9f6e348e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4ac6f453f097afcdfc66e04aae1b733a9f6e348e"}]},"branch":"refs/heads/master"},"136ccacbac9012438985530137148ccc2d9fc448":{"kind":"REWORK","_number":4,"created":"2020-08-05 10:36:30.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/4"}}},"commit":{"parents":[{"commit":"d3229db6c98575b0e7607225961aefedc957e82c","subject":"Merge \"Update version of hacking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d3229db6c98575b0e7607225961aefedc957e82c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 10:36:05.000000000","tz":120},"subject":"Simplify the Manager driver","message":"Simplify the Manager driver\n\nUse a get_manager call instead of uuid/name calls that transform\nidentities. Simplify get_manager to not fetch data it ignores anyway.\nMove the (broken) systems and chassis listing logic to the statis\nmanager driver, so that it can be replaced by something working.\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/136ccacbac9012438985530137148ccc2d9fc448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/136ccacbac9012438985530137148ccc2d9fc448"}]},"branch":"refs/heads/master"},"e23bcccdcd860b922262c341e624437c657c0b00":{"kind":"REWORK","_number":5,"created":"2020-08-05 11:36:47.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/5"}}},"commit":{"parents":[{"commit":"4120f89a3bbb3a0577125c468c840359bc0ab61b","subject":"Log API errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4120f89a3bbb3a0577125c468c840359bc0ab61b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 11:36:04.000000000","tz":120},"subject":"Simplify the Manager driver","message":"Simplify the Manager driver\n\nUse a get_manager call instead of uuid/name calls that transform\nidentities. Simplify get_manager to not fetch data it ignores anyway.\nMove the (broken) systems and chassis listing logic to the statis\nmanager driver, so that it can be replaced by something working.\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e23bcccdcd860b922262c341e624437c657c0b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e23bcccdcd860b922262c341e624437c657c0b00"}]},"branch":"refs/heads/master"},"4aa1d3164ba5e6e8de252660f07b52ae4a38dc0a":{"kind":"REWORK","_number":6,"created":"2020-08-05 11:41:18.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/6"}}},"commit":{"parents":[{"commit":"4120f89a3bbb3a0577125c468c840359bc0ab61b","subject":"Log API errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4120f89a3bbb3a0577125c468c840359bc0ab61b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 11:40:58.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4aa1d3164ba5e6e8de252660f07b52ae4a38dc0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4aa1d3164ba5e6e8de252660f07b52ae4a38dc0a"}]},"branch":"refs/heads/master"},"6b155f4be2e3d5b1e9cc138bf537b7742250d960":{"kind":"REWORK","_number":7,"created":"2020-08-05 11:43:29.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/7"}}},"commit":{"parents":[{"commit":"4120f89a3bbb3a0577125c468c840359bc0ab61b","subject":"Log API errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4120f89a3bbb3a0577125c468c840359bc0ab61b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 11:43:17.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6b155f4be2e3d5b1e9cc138bf537b7742250d960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6b155f4be2e3d5b1e9cc138bf537b7742250d960"}]},"branch":"refs/heads/master"},"574a95e88511ffcd2dc27ee999b6e6c1c121e18e":{"kind":"REWORK","_number":8,"created":"2020-08-05 12:37:28.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/8"}}},"commit":{"parents":[{"commit":"4120f89a3bbb3a0577125c468c840359bc0ab61b","subject":"Log API errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4120f89a3bbb3a0577125c468c840359bc0ab61b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 12:37:17.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/574a95e88511ffcd2dc27ee999b6e6c1c121e18e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/574a95e88511ffcd2dc27ee999b6e6c1c121e18e"}]},"branch":"refs/heads/master"},"cc7dff5387eaa4c160a8fb548b957f899ebc811a":{"kind":"REWORK","_number":9,"created":"2020-08-05 13:14:32.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/9"}}},"commit":{"parents":[{"commit":"4120f89a3bbb3a0577125c468c840359bc0ab61b","subject":"Log API errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4120f89a3bbb3a0577125c468c840359bc0ab61b"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 13:14:22.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/cc7dff5387eaa4c160a8fb548b957f899ebc811a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/cc7dff5387eaa4c160a8fb548b957f899ebc811a"}]},"branch":"refs/heads/master"},"024c3d8b572fc4a2e86be59ee3fdcef5bc0fa0dd":{"kind":"REWORK","_number":10,"created":"2020-08-06 07:43:27.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/10"}}},"commit":{"parents":[{"commit":"fab49ebab581620b94e5c46e0101ad78feea5afa","subject":"Log API errors for easier debugging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fab49ebab581620b94e5c46e0101ad78feea5afa"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-06 07:43:13.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/024c3d8b572fc4a2e86be59ee3fdcef5bc0fa0dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/024c3d8b572fc4a2e86be59ee3fdcef5bc0fa0dd"}]},"branch":"refs/heads/master"},"b542eed95697af83354aa2d02a73f51d4478708b":{"kind":"REWORK","_number":11,"created":"2020-08-06 08:42:41.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/11"}}},"commit":{"parents":[{"commit":"fab49ebab581620b94e5c46e0101ad78feea5afa","subject":"Log API errors for easier debugging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fab49ebab581620b94e5c46e0101ad78feea5afa"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-06 08:42:32.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b542eed95697af83354aa2d02a73f51d4478708b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b542eed95697af83354aa2d02a73f51d4478708b"}]},"branch":"refs/heads/master"},"78f983e9c7f7659583a48ea78c7681669ab5685f":{"kind":"REWORK","_number":12,"created":"2020-08-06 09:42:35.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/12"}}},"commit":{"parents":[{"commit":"fab49ebab581620b94e5c46e0101ad78feea5afa","subject":"Log API errors for easier debugging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fab49ebab581620b94e5c46e0101ad78feea5afa"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-06 09:42:26.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/78f983e9c7f7659583a48ea78c7681669ab5685f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/78f983e9c7f7659583a48ea78c7681669ab5685f"}]},"branch":"refs/heads/master"},"b3bbb5b108216e39e5d6a98ae34e3ec38b568cf5":{"kind":"REWORK","_number":13,"created":"2020-08-06 10:50:31.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/744862/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/62/744862/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/62/744862/13"}}},"commit":{"parents":[{"commit":"dba36e54c52276b0f33885cb814ca5331028cd39","subject":"Merge \"Use property plus abstractmethod for abstractproperty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/dba36e54c52276b0f33885cb814ca5331028cd39"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-05 09:01:55.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-06 10:50:16.000000000","tz":120},"subject":"Replace the static Manager driver with a new fake one","message":"Replace the static Manager driver with a new fake one\n\nRefactors the Manager interface to provide a single get_manager call\ninstead of uuid/name calls that transform identities.\n\nAdd new calls to list systems and chassis per manager.\n\nDo not use jinja for rendering JSON, it\u0027s redundant, just use Python.\n\nFinally, provide a manager driver that ties a manager to a system\nin a 1-1 relationship. This fixes the current issue that one manager\nmanagers all systems, and thus a virtual media is always attached\nto all systems.\n\nChange-Id: I46dd794750c182572d37bab3ebfcd608ceb1bb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b3bbb5b108216e39e5d6a98ae34e3ec38b568cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b3bbb5b108216e39e5d6a98ae34e3ec38b568cf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
