)]}'
{"id":"openstack%2Fsushy-tools~766071","triplet_id":"openstack%2Fsushy-tools~master~I5d1657a403a70d0e54998ff2f6966eb1f3fc41be","project":"openstack/sushy-tools","branch":"master","topic":"pip-resolver","hashtags":[],"change_id":"I5d1657a403a70d0e54998ff2f6966eb1f3fc41be","subject":"Fix l-c for sushy-tools","status":"MERGED","created":"2020-12-08 18:45:07.000000000","updated":"2020-12-08 20:47:19.000000000","submitted":"2020-12-08 20:45:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766071-pip-resolver","meta_rev_id":"8067e7726bad0748ecbadedfdaf2c3da75e6ac8f","_number":766071,"virtual_id_number":766071,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-08 20:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-12-08 18:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2020-12-08 18:50:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 18:50:55.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-12-08 19:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"901cd092bed22a3f9a9acb6e5e360ac3aa9e12b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-12-08 18:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40a3503728666d817e5b7b47b71b6d9b8caba92c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-08 18:50:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f56befebb199588f9f52c60aea5c1a70665bdad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 19:48:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6db41e51b14eab95a1171a0088704c : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c26374224344789b841604105e0f7ee : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd57ae390fe5433093ef6518aa19a4ba : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f7f0b97c529438b9e44a7db9c2b4736 : SUCCESS in 5m 28s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3396503f6484e2f96285fcbe8dabdf0 : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc4db2e67569426ab8c120a82c8b949f : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3299e24a109d4749bf473635dbfc1bd7 : SUCCESS in 7m 31s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a26e3ec347c0486c930d79248d3df281 : SUCCESS in 44m 38s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f31f52b227204ead8f6b76610ee29f18 : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c5c508c73640b3622723dd1f9402fc00608ab919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 19:48:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9920acec4002de269cf8f7443560ac6b19a52eed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 20:45:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54fb08c1567c4f398dd7c291b9012226 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74145b4f3dfa43c8b8a1297ef062d158 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0e39fc907e846b1af23a79a779798de : SUCCESS in 4m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c96f19ee26734c3aa32042e47b0a6ddb : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f52a0cc4e53944289e91d1a952c4c5c0 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53cb588afb154936af60757846439695 : SUCCESS in 8m 20s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f81797b0b1af4bf298947ae75e6054d7 : SUCCESS in 54m 53s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0614bf6166eb4114b845cb7498af6414 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"226ff50163c9ad29d703fe5463dc63075392dab9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 20:45:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8067e7726bad0748ecbadedfdaf2c3da75e6ac8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 20:47:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1234b57b05b4457bb451a21402b224ce : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dce7f09b9d92ca10b576c1b76db1cfb461db9b4","revisions":{"8dce7f09b9d92ca10b576c1b76db1cfb461db9b4":{"kind":"REWORK","_number":1,"created":"2020-12-08 18:45:07.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/71/766071/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/71/766071/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/71/766071/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/71/766071/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/71/766071/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/71/766071/1"}}},"commit":{"parents":[{"commit":"e7dbba93f19b1b7aaf936896b3be5ede2a024cab","subject":"Allow setting debug mode from command line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e7dbba93f19b1b7aaf936896b3be5ede2a024cab"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2020-12-08 18:43:06.000000000","tz":60},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2020-12-08 18:45:00.000000000","tz":60},"subject":"Fix l-c for sushy-tools","message":"Fix l-c for sushy-tools\n\n* bump docutils required by sphinx\n\nChange-Id: I5d1657a403a70d0e54998ff2f6966eb1f3fc41be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8dce7f09b9d92ca10b576c1b76db1cfb461db9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8dce7f09b9d92ca10b576c1b76db1cfb461db9b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
