)]}'
{"id":"openstack%2Fsushy-tools~790361","triplet_id":"openstack%2Fsushy-tools~master~Id63ef9076123a604c1e4ffaa48f9b6acd12b3517","project":"openstack/sushy-tools","branch":"master","topic":"bios-registry","hashtags":["ironic-week-prio"],"change_id":"Id63ef9076123a604c1e4ffaa48f9b6acd12b3517","subject":"Add support for returning message and attribute registries","status":"MERGED","created":"2021-05-09 23:39:13.000000000","updated":"2021-06-01 10:07:13.000000000","submitted":"2021-06-01 10:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790361-bios-registry","meta_rev_id":"4ea2b5c79730a56ff902670db8c4e15e7cdb36b0","_number":790361,"virtual_id_number":790361,"owner":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 10:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":1,"date":"2021-05-17 13:58:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-05-26 11:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-01 09:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-01 09:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 00:36:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 10:44:12.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2021-05-13 17:29:27.000000000","updated_by":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-05-26 11:26:24.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-06-01 09:07:35.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"705a790df2efdd98c6c9f6e04897cd62e571dd01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-09 23:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d583b72daaa8b6fefa761d956b247558e629ff5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 00:36:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f2b340af484e2caffedcc3ff18aa5a : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cfd5b30529b41da9d632a7f8d4f36f6 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8967bb92e9594dada1ce2ada91276699 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf1f0d57fa164f2e97d87f7d6c572745 : SUCCESS in 4m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a37283ed6f294895b24093e08c450dc1 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7cc5c0988f640a997e046df47252c5a : SUCCESS in 5m 51s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0f42f53bd51a4f6d9800edc7daf9a873 : SUCCESS in 51m 03s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/19dd032cf1294bf7b143a3854b0a10ed : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0007aac6b8616d2a4b3ccc76068070f7c0dc9b8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-11 14:36:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb54a723012dbc04066f0af5a86fbfe421dfda2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:30:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/093a8ca614714a79b83ba27d023c0735 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1648e00af5d439a8b56370e12ec4a97 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97ae065ce95b40a69c2da5cd2f87b7e0 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d615904e74d4a04bbfd630f4cb9eca2 : SUCCESS in 4m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bd5eb76db3643c28de99578f2e547e7 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f656f0d46e4093b549934f8a889dd9 : SUCCESS in 7m 06s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/46c7ec97cbc14bc39b6ac5d483ad375e : SUCCESS in 47m 50s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/14bc54fa4a9d4abcbfe8ad0fbac1eb18 : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e17dd878908807c9aa853d10237df224c0d9cdd7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-12 10:44:12.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nhad a quick read, just a couple of nits","accounts_in_message":[],"_revision_number":2},{"id":"d5637100b38fcbed67913b2a9ca77d045ac1db17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-13 16:26:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e166388243a971fd87173327108e592d980f4dcc","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-13 16:34:05.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nThanks Riccardo!","accounts_in_message":[],"_revision_number":3},{"id":"e21b5a734175fb622d20db0bb29e1d53073e4fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 17:22:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0683d09d428548ef90613456cb1669fd : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41c4296b432a415aacc92b38fe4351c3 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4661af1dfb3748f4bd5082eb4c2c922a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c91d632cc4e484d8594936d0ca405e9 : SUCCESS in 3m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7daf3f6b281849eaa46b0af23bb65ca0 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1f06120a4243b6ac95fe6ac54551bd : SUCCESS in 5m 31s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8c2fd8322b684720a34e25f9d0094638 : SUCCESS in 48m 35s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bcd1a44c96664506bc654baac775f0a1 : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6c9681436c7b1a1966950f8d049633666cb78c12","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-14 15:21:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)\n\nthanks for the patch, need to verify if \u003cidentity\u003e is missing from 1 route.","accounts_in_message":[],"_revision_number":3},{"id":"c8e1d570be2f2014260bd324298709d0039e1be8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-14 17:17:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fe001f7b25774e2eb5fb411e2c6a473612862950","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-14 17:20:44.000000000","message":"Patch Set 4:\n\n(8 comments)\n\nThanks a lot for the review Aija!  Comments addressed below.","accounts_in_message":[],"_revision_number":4},{"id":"c78e68818c40d48d674bc7195d1a9b790f0ef250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:09:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef55c1dfb6494751951568923f4b7f44 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ad86920857d468a83e8684eb908d0b0 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58d9cc4e380346c0acb6ada083b24a17 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1205efa4ab46407da82bc5aba330febc : SUCCESS in 4m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3077197480a406c8a7c011734a805e0 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde9f46318174dcfabf4f4c2277a67ef : SUCCESS in 7m 58s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a806b3d58fa54791985d3d250c460b09 : SUCCESS in 42m 29s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c4c02d9ff0c840bb9fac384fcd5ce19e : SUCCESS in 50m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2b99250891b0fb77052b726232788415ba475c25","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-17 07:37:13.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)\n\nthanks for updates","accounts_in_message":[],"_revision_number":4},{"id":"8f88d4af9d789c0e3f169c72892edee2c5889d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-17 11:21:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b2777ab751c65fd0b9dfb44d3bc49684517538f2","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-17 11:22:26.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5d2ba0ed458b39ac87e8a696f507ee0125f40d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 12:07:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59414bcd314e45d4b2a9bf4bd2c44433 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b81c0146286b4d4f8693806884d34946 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44e48ddc65584ae19e7a14d640d030d2 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90d8bf8780b047dea5736c74ed0fca20 : SUCCESS in 5m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77232aa510994d3cabb2adb3ce017fde : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b8e31e52954d3ea65e05d1a65015b5 : SUCCESS in 5m 44s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ced30dde34f0428b9f9fedd03c1ce231 : SUCCESS in 38m 24s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1ed4e42b61d64fdabd17ac05cc91605b : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":5},{"id":"e105a9f36df4dd543cf6edd0c9c2dadfc51c3979","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-17 13:58:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6c71e91b3001a430567e37d06beaefe83ffffc81","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-26 11:26:24.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"62c6e3beeaede1e3fec293cc9fa63d6a1edd5531","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-31 15:19:50.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":5},{"id":"449feaae0c67359d29992c6f128037a41b439665","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-06-01 09:07:35.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2c27deb253dbc068dcaa7665ec116e74f56821bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 09:07:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0a4eede39d226c6df7a2b318d188aa5cbd5aab28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:05:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e6b0437489404e91f60d82ba071951 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0db39ee03db248389423f8b4a5b5c2b4 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af70e0c1c3b24eccb4a637e12ff678eb : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97c32d8efe140f9981bd69f7e904eb5 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b4ea9de50b41459ba9177b7efa736f : SUCCESS in 6m 05s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/00b4638fd295412db66fbbb185699681 : SUCCESS in 51m 28s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2246f3c88045432fb8ab805b08fdf916 : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":5},{"id":"a32ca97943f45d70bd8b406e7396e9e0be188786","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:05:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4ea2b5c79730a56ff902670db8c4e15e7cdb36b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:07:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69f8674e88324d82998b4a000e94b6bb : SUCCESS in 1m 05s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89872f3a041b4d9595ec347c3e0b51c2 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"de31f7b5b2579be2e6468f522147016393fff94d","revisions":{"38a2606e2f8a63b0c22a6ae7bd957493dddf1cd2":{"kind":"REWORK","_number":1,"created":"2021-05-09 23:39:13.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/61/790361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/61/790361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/1"}}},"commit":{"parents":[{"commit":"17fd2a01d3984a6c0715cade56bd9259fd47dcdc","subject":"Merge \"Return an integer value for Cores and Threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/17fd2a01d3984a6c0715cade56bd9259fd47dcdc"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-09 23:37:33.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-09 23:39:12.000000000","tz":-240},"subject":"Add support for returning message and attribute registries","message":"Add support for returning message and attribute registries\n\nAdded support for returning message and attribute (particularly BIOS\nattribute) registries. Also added two additional BIOS settings -\n\u0027SecureBootStatus\u0027 and \u0027SerialNumber\u0027. \u0027SecureBootStatus\u0027 has \u0027ReadOnly\u0027\nset to true and \u0027SecureBootStatus\u0027 has \u0027IsSystemUniqueProperty\u0027 in the\nregistry.\n\nChange-Id: Id63ef9076123a604c1e4ffaa48f9b6acd12b3517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/38a2606e2f8a63b0c22a6ae7bd957493dddf1cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/38a2606e2f8a63b0c22a6ae7bd957493dddf1cd2"}]},"branch":"refs/heads/master"},"b469c6a4ced42551e7d6b6d8722ce9bc660a9afa":{"kind":"REWORK","_number":2,"created":"2021-05-11 14:36:22.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/61/790361/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/61/790361/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/2"}}},"commit":{"parents":[{"commit":"17fd2a01d3984a6c0715cade56bd9259fd47dcdc","subject":"Merge \"Return an integer value for Cores and Threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/17fd2a01d3984a6c0715cade56bd9259fd47dcdc"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-09 23:37:33.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-11 14:36:11.000000000","tz":-240},"subject":"Add support for returning message and attribute registries","message":"Add support for returning message and attribute registries\n\nAdded support for returning message and attribute (particularly BIOS\nattribute) registries. Also added two additional BIOS settings -\n\u0027SecureBootStatus\u0027 and \u0027SerialNumber\u0027. \u0027SecureBootStatus\u0027 has \u0027ReadOnly\u0027\nset to true and \u0027SecureBootStatus\u0027 has \u0027IsSystemUniqueProperty\u0027 in the\nregistry.\n\nChange-Id: Id63ef9076123a604c1e4ffaa48f9b6acd12b3517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b469c6a4ced42551e7d6b6d8722ce9bc660a9afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b469c6a4ced42551e7d6b6d8722ce9bc660a9afa"}]},"branch":"refs/heads/master"},"8a376276f44c5d03800abdf3b7bf95b647f6ca01":{"kind":"REWORK","_number":3,"created":"2021-05-13 16:26:26.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/61/790361/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/61/790361/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/3"}}},"commit":{"parents":[{"commit":"17fd2a01d3984a6c0715cade56bd9259fd47dcdc","subject":"Merge \"Return an integer value for Cores and Threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/17fd2a01d3984a6c0715cade56bd9259fd47dcdc"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-09 23:37:33.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-13 16:26:16.000000000","tz":-240},"subject":"Add support for returning message and attribute registries","message":"Add support for returning message and attribute registries\n\nAdded support for returning message and attribute (particularly BIOS\nattribute) registries. Also added two additional BIOS settings -\n\u0027SecureBootStatus\u0027 and \u0027SerialNumber\u0027. \u0027SecureBootStatus\u0027 has \u0027ReadOnly\u0027\nset to true and \u0027SecureBootStatus\u0027 has \u0027IsSystemUniqueProperty\u0027 in the\nregistry.\n\nChange-Id: Id63ef9076123a604c1e4ffaa48f9b6acd12b3517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8a376276f44c5d03800abdf3b7bf95b647f6ca01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8a376276f44c5d03800abdf3b7bf95b647f6ca01"}]},"branch":"refs/heads/master"},"ca4f340b18a99a061c968050d275b53f83595cdd":{"kind":"REWORK","_number":4,"created":"2021-05-14 17:17:28.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/61/790361/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/61/790361/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/4"}}},"commit":{"parents":[{"commit":"17fd2a01d3984a6c0715cade56bd9259fd47dcdc","subject":"Merge \"Return an integer value for Cores and Threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/17fd2a01d3984a6c0715cade56bd9259fd47dcdc"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-09 23:37:33.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-14 17:16:32.000000000","tz":-240},"subject":"Add support for returning message and attribute registries","message":"Add support for returning message and attribute registries\n\nAdded support for returning message and attribute (particularly BIOS\nattribute) registries. Also added two additional BIOS settings -\n\u0027SecureBootStatus\u0027 and \u0027SerialNumber\u0027. \u0027SecureBootStatus\u0027 has \u0027ReadOnly\u0027\nset to true and \u0027SecureBootStatus\u0027 has \u0027IsSystemUniqueProperty\u0027 in the\nregistry.\n\nChange-Id: Id63ef9076123a604c1e4ffaa48f9b6acd12b3517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ca4f340b18a99a061c968050d275b53f83595cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ca4f340b18a99a061c968050d275b53f83595cdd"}]},"branch":"refs/heads/master"},"de31f7b5b2579be2e6468f522147016393fff94d":{"kind":"REWORK","_number":5,"created":"2021-05-17 11:21:51.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/61/790361/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/61/790361/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/61/790361/5"}}},"commit":{"parents":[{"commit":"17fd2a01d3984a6c0715cade56bd9259fd47dcdc","subject":"Merge \"Return an integer value for Cores and Threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/17fd2a01d3984a6c0715cade56bd9259fd47dcdc"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-09 23:37:33.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-05-17 11:21:42.000000000","tz":-240},"subject":"Add support for returning message and attribute registries","message":"Add support for returning message and attribute registries\n\nAdded support for returning message and attribute (particularly BIOS\nattribute) registries. Also added two additional BIOS settings -\n\u0027SecureBootStatus\u0027 and \u0027SerialNumber\u0027. \u0027SecureBootStatus\u0027 has \u0027ReadOnly\u0027\nset to true and \u0027SecureBootStatus\u0027 has \u0027IsSystemUniqueProperty\u0027 in the\nregistry.\n\nChange-Id: Id63ef9076123a604c1e4ffaa48f9b6acd12b3517\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/de31f7b5b2579be2e6468f522147016393fff94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/de31f7b5b2579be2e6468f522147016393fff94d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
