)]}'
{"id":"openstack%2Fsushy-tools~836795","triplet_id":"openstack%2Fsushy-tools~master~I5c9eae02c00843b265fcaf0d041130256d65500f","project":"openstack/sushy-tools","branch":"master","topic":"image","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-05-10 10:00:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c9eae02c00843b265fcaf0d041130256d65500f","subject":"vmedia: keep the original URL in Image","status":"MERGED","created":"2022-04-06 10:33:52.000000000","updated":"2022-05-10 10:01:42.000000000","submitted":"2022-05-10 10:00:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836795-image","meta_rev_id":"06f35bb2d95cac4e07c1193411075b1f3dcd24d6","_number":836795,"virtual_id_number":836795,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 10:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2022-05-10 09:53:51.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2022-04-26 14:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2022-05-10 09:53:51.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 12:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 14:57:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-05-10 09:53:51.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"}],"messages":[{"id":"5995b049808159cfe88e6b9e447ed8954c08d38c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 10:33:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78b6d078e1a1714592d9bb98894fd70c67e3a28d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 10:58:09.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1ede563f861aa4f3f042eb43d4db8abdf43d489e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:14:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a8bc6dee40543bcbc4505fb0d56741f : FAILURE in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1435936dd3384fd3b1db672d080172f5 : SUCCESS in 5m 34s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ac4e54c91c594f3d839aaef7441ba176 : SUCCESS in 56m 24s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f52e9575f0cb400db969c1ab4447aaac : SUCCESS in 1h 01m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8645e8af1e4b5d90dc2907f9c88f3d : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cce5cb6ffb3a42e98dd314d64b0d0836 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec72f4a4e04f4def93fef98a81ad5f5f : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99b91d22123041c0b6209e4b05c3e5f8 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7760494ded5423aa8207f4d410a9e05 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"20e8f244e5df4e9493a9e4941e2272ddb3f33da0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-26 14:54:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9fb33532b38cf707c895ab6c7fae6bc3a470169f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-26 14:57:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e755ae9ba204e2158786bcb7fec574bd2858ee60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:52:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaf0888657184c9d8c34c0a6e1f4f475 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728a1c7dc69344f5a46d4082d84fbdff : SUCCESS in 6m 31s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b317dc662f6c4f6babd1c1d8b056d94e : SUCCESS in 49m 13s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bff28fdaa4594a48ac759d8add8fc25c : SUCCESS in 56m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6483f75734742c491bd773bbdbe02d3 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d7135e2909347e184f14926c2b4ae98 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5a0d470c84943619ebc55982097ccb5 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a20b915783ec48adafdd0dec568266ea : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0209fde91b54a56ab11a435143486d5 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"6dfdac54640fce3a7406cbd2b3bfd627d3a56c0e","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2022-05-10 09:53:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56d4362871d1b471de5ffb94b6870e9c5c031c5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:54:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e0a98b464bed17e997071354c30382d7ed998d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:00:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d458d66a625b41cd8c7f6bd603808f44 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e1b99f59c994a61aa99ad926377a9b7 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c04251cbafc7499685a226096fb82cda : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ca9f2228484ad78818c13c266acddc : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/361fc74f9c74432c8c2d28611cdfbcff : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ec59fd03b73b9264af8cc87de77f3424e1ea9089","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:00:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"06f35bb2d95cac4e07c1193411075b1f3dcd24d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:01:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e727c1608ee45bea3dedd15a6f41e0c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c34ffbd402071eaf8d303b0edb487ab3632f273b","revisions":{"5bb6667fd27269aefcc9371bec5846c01dd902c6":{"kind":"REWORK","_number":1,"created":"2022-04-06 10:33:52.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/95/836795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/836795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/1"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:33:11.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:33:11.000000000","tz":120},"subject":"vmedia: keep the original URL in Image","message":"vmedia: keep the original URL in Image\n\nImage must be the URL that the user provided, not our local path.\nUse ImageName for the local path instead.\n\nChange-Id: I5c9eae02c00843b265fcaf0d041130256d65500f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5bb6667fd27269aefcc9371bec5846c01dd902c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5bb6667fd27269aefcc9371bec5846c01dd902c6"}]},"branch":"refs/heads/master"},"db0928c40a435d2ea2be0ccae20f33ad3fc6f3a7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-06 10:58:09.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/95/836795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/836795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/2"}}},"commit":{"parents":[{"commit":"8eea6be3858ede1678fc57ec5b357ca3604cb45f","subject":"Fix the unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8eea6be3858ede1678fc57ec5b357ca3604cb45f"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:33:11.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:57:31.000000000","tz":120},"subject":"vmedia: keep the original URL in Image","message":"vmedia: keep the original URL in Image\n\nImage must be the URL that the user provided, not our local path.\nUse ImageName for the local path instead.\n\nChange-Id: I5c9eae02c00843b265fcaf0d041130256d65500f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/db0928c40a435d2ea2be0ccae20f33ad3fc6f3a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/db0928c40a435d2ea2be0ccae20f33ad3fc6f3a7"}]},"branch":"refs/heads/master"},"c34ffbd402071eaf8d303b0edb487ab3632f273b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-26 14:54:13.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/95/836795/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/95/836795/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/95/836795/3"}}},"commit":{"parents":[{"commit":"8896740bbe2af6244796cc46a873b70048c636a6","subject":"Fix the CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8896740bbe2af6244796cc46a873b70048c636a6"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:33:11.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-26 14:53:57.000000000","tz":120},"subject":"vmedia: keep the original URL in Image","message":"vmedia: keep the original URL in Image\n\nImage must be the URL that the user provided, not our local path.\nUse ImageName for the local path instead.\n\nChange-Id: I5c9eae02c00843b265fcaf0d041130256d65500f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c34ffbd402071eaf8d303b0edb487ab3632f273b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c34ffbd402071eaf8d303b0edb487ab3632f273b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
