)]}'
{"id":"openstack%2Fsushy-tools~836801","triplet_id":"openstack%2Fsushy-tools~master~I5bad754ef7b1bec6986c2b4293a38226c5a261af","project":"openstack/sushy-tools","branch":"master","topic":"ci","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-04-26 15:48:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5bad754ef7b1bec6986c2b4293a38226c5a261af","subject":"Fix the CI","status":"MERGED","created":"2022-04-06 10:57:23.000000000","updated":"2022-04-26 15:52:39.000000000","submitted":"2022-04-26 15:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836801-ci","meta_rev_id":"3c0367a8dc7d5c0b4c122f2a11ec3ded430b0df0","_number":836801,"virtual_id_number":836801,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 15:48:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2022-04-26 14:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2022-04-26 09:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2022-04-26 14:44:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 11:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 13:19:42.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"CC"},{"updated":"2022-04-26 09:51:11.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-04-26 14:44:52.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"2dd1b5ab686c78ecc4cc2b1b0e1b2398b49ee1df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 10:57:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5295d22afee90576331c01e508082e32bd3e0826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 11:11:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29696b8bf0ee4cbea01ccb1ce551751f : FAILURE in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ade25e7413d499797011406347ae80e : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b743d42b2c4bcd9c3e70341c04329e : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2427bd8fbdd842a49da4c05a52b020ff : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7faba668796f4e0ca7ee54755408edf9 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa07bdea8c1049a49af7a8e1954a3c80 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be0c3776356b473ebe75eee47efa174f : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a7126f90903d05471a407e674b677ed80f4fce1a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-06 13:19:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c30612b5ebbe477fbf1c570d54d04428104801b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 14:11:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9eca204a66bc6e2176a2335bd82c65d456d37c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 15:08:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cbe86c152449451bb31ec787ce2314a3 : SUCCESS in 2m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd17da4e3dfe49199cb0cde9810955c1 : FAILURE in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228366155edf4e18a76b350843d80e55 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72a3dd4370e2447bb35df254c364fbcb : SUCCESS in 3m 41s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2e506cbb1ad1477b82856da602a0b1c5 : SUCCESS in 47m 10s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e89971ff862940d38c4239e3ea8e424a : SUCCESS in 53m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d769a66c5e5425a909559fdb06c0f66 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a1e258cea34e5dab4343c23f0280af : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21f2cf654a80428d858ab5da509f04fa : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8c4099a2b5d4176a579f398eeadafd8 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d79b57c5a714bffbca44edb44627163 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"6e2a400c8bf0fde86151b58cb9cfca76524c06de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 15:11:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"49c1b7571832bd0ac174af0fdd20db8dc4711cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 15:34:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7246ebbf97b23ccf440e45e28e89ffe66a0f5df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 15:46:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"64da20993a1e30b1f7de1fb9db8dcfbebbe540ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-06 16:01:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"86ebd8d9699c740aa6f8ba1feb92bec200d5834b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-06 16:38:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"359f8808eaed23e223e522579d07cbb3130ac275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 17:03:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/14e253917c354c5e9368ae5f57968523 : SUCCESS in 2m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7d151d919724f23b25d7ed09f0d234d : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5294f70693744317b49071a5fbc0a51a : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c305e7fe9e7f4813b8354e79bae4ba53 : SUCCESS in 3m 50s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/77213370153444d982fec45d84821728 : SUCCESS in 54m 53s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6b03e8107e1f493f8a92a512be45eae8 : SUCCESS in 56m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba75445b43ad46d0967a9f285b328ff4 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd3affe0dc574a1bab01102385814478 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a9b8542d74d4affaa8a54704e5b5fc1 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1bb689dc4e4469b86d6637d8769efc3 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ead08846ae494cc2a7f033783fa0ca20 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"429f47728a597cc5725787a382f659b8935569bb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-07 11:02:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6fa4f7e2594c068cc50e83ed7368226c3de5f294","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-07 12:57:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c5a625b34c8176a79358a58ed8a1d49dcfbdc173","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-07 15:47:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6ff271e39a35ed88ef01f09eca0bacbfee22584a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-04-26 07:54:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4dc0929ca549c543297972743293832c92737757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 08:55:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0fcdfba5b4834fa586cf423eefd3f88a : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a137c7259494291bcb509a24d2d0679 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8fcb1c13d5433b8bfb07bda284f338 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38795a82bff54c98bd792d118399dda1 : SUCCESS in 3m 37s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b3798fb4dbec479fa7d39517799dea7d : SUCCESS in 53m 10s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bf358f5521a346d19b7ca4cedaa0f04e : SUCCESS in 56m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e82b2c91a140afa56fafddf6a9d005 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d992c59de035496d9b22fda8076adb1f : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80e9f8830c2a4ce293b08978f9837e7a : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6686e6adfb82486794ca98f3fb262dec : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8197a2e23dd4bd7bc91b26776e74253 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":7},{"id":"aa0bd5479b2ce6c7cf2121fbc96e0485b7cb2e4f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-26 09:51:11.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"adb961ba8f9633a2a6b1d88c54a3b270d499d377","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-04-26 14:44:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"45105910ab28139818ac8aa2a5b016d9c57ccf9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 14:45:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6b0911ec047097323fdb426c7079bd8ce6d5e211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:48:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b26cf1fe39104e27b4f460dd0f73222b : SUCCESS in 2m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6616e2e9d710456ebf065e2faf9e30b0 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7fdb3c04df4b719228610b25b55d88 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a56ac756bf46ca8fd3b219bfc6a4b8 : SUCCESS in 3m 46s\n- sushy-tools-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b84dd27338674c50a08cc07050f0b510 : SUCCESS in 57m 08s\n- sushy-tools-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e449132beab94ea5b744232a53985fed : SUCCESS in 52m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca72292bb042469e9e0bf728ed13a42e : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d54d1a3474ae4f7c9f50068c128e2e72 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/983928f200cb4534906de7473d9efc5d : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3791beae92f4dffbe8385756118e6c2 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/155fba50c2a041b0aba5580d249f121e : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":7},{"id":"47828ab181601798e903240963c6979b820e1435","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:48:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"3c0367a8dc7d5c0b4c122f2a11ec3ded430b0df0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 15:52:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75217d2fbc854a1fad9ee89cad4ab283 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4675c113993469fa2b57362e2ec11d8 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8896740bbe2af6244796cc46a873b70048c636a6","revisions":{"8eea6be3858ede1678fc57ec5b357ca3604cb45f":{"kind":"REWORK","_number":1,"created":"2022-04-06 10:57:23.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/1"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"subject":"Fix the unit tests","message":"Fix the unit tests\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8eea6be3858ede1678fc57ec5b357ca3604cb45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8eea6be3858ede1678fc57ec5b357ca3604cb45f"}]},"branch":"refs/heads/master"},"01d79a6b2bb1517628a6bae307340486fd5a0b03":{"kind":"REWORK","_number":2,"created":"2022-04-06 14:11:50.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/2"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 14:10:58.000000000","tz":120},"subject":"Fix the CI","message":"Fix the CI\n\nLower constraints do not cap all dependencies, causing incompatibilities.\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/01d79a6b2bb1517628a6bae307340486fd5a0b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/01d79a6b2bb1517628a6bae307340486fd5a0b03"}]},"branch":"refs/heads/master"},"12fd572a78161b1693a29e3520a9800a4957f05b":{"kind":"REWORK","_number":3,"created":"2022-04-06 15:11:39.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/3"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 15:11:08.000000000","tz":120},"subject":"Fix the CI","message":"Fix the CI\n\nLower constraints do not cap all dependencies, causing incompatibilities.\nUpdate Jinja2 to 2.11.3 (the last pre-3.0 version) for Flask.\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/12fd572a78161b1693a29e3520a9800a4957f05b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/12fd572a78161b1693a29e3520a9800a4957f05b"}]},"branch":"refs/heads/master"},"0a29b9359756316df0c7a78f8751d8490148eadc":{"kind":"REWORK","_number":4,"created":"2022-04-06 15:34:03.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/4"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 15:33:31.000000000","tz":120},"subject":"Fix the CI","message":"Fix the CI\n\nLower constraints do not cap all dependencies, causing incompatibilities.\nUpdate Jinja2 to 2.11.3 (the last pre-3.0 version) and downgrade\nMarkupSafe to 2.0.1 for Flask.\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0a29b9359756316df0c7a78f8751d8490148eadc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0a29b9359756316df0c7a78f8751d8490148eadc"}]},"branch":"refs/heads/master"},"ba66c9b7829511bf3f838769898f09928b9e8c3f":{"kind":"REWORK","_number":5,"created":"2022-04-06 15:46:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/5"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 15:45:37.000000000","tz":120},"subject":"Fix the CI","message":"Fix the CI\n\nLower constraints do not cap all dependencies, causing incompatibilities.\nUpdate Jinja2 to 2.11.3 (the last pre-3.0 version) and downgrade\nMarkupSafe and itsdangerous for Flask.\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ba66c9b7829511bf3f838769898f09928b9e8c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/ba66c9b7829511bf3f838769898f09928b9e8c3f"}]},"branch":"refs/heads/master"},"5d1620500402544339d785012db5cc149b4c0deb":{"kind":"REWORK","_number":6,"created":"2022-04-06 16:01:17.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/6"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 16:01:06.000000000","tz":120},"subject":"Fix the CI","message":"Fix the CI\n\nLower constraints do not cap all dependencies, causing incompatibilities.\nUpdate Jinja2 to 2.11.3 (the last pre-3.0 version) and downgrade\nMarkupSafe, Werkzeug and itsdangerous for Flask.\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5d1620500402544339d785012db5cc149b4c0deb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5d1620500402544339d785012db5cc149b4c0deb"}]},"branch":"refs/heads/master"},"8896740bbe2af6244796cc46a873b70048c636a6":{"kind":"REWORK","_number":7,"created":"2022-04-26 07:54:41.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/01/836801/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/01/836801/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/01/836801/7"}}},"commit":{"parents":[{"commit":"1553bf31823c2a36a90830de8e95e85d96af78dc","subject":"Accept non-string types for BIOS settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/1553bf31823c2a36a90830de8e95e85d96af78dc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-06 10:56:51.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-04-26 07:54:13.000000000","tz":120},"subject":"Fix the CI","message":"Fix the CI\n\nLower constraints do not cap all dependencies, causing incompatibilities.\nUpdate Jinja2 to 2.11.3 (the last pre-3.0 version) and downgrade\nMarkupSafe, Werkzeug and itsdangerous for Flask.\n\nSomething must have changed in one of the dependencies, and the Location\nheader is no longer returning the host in the URL.\n\nRemove subunit and testscenarios, we don\u0027t use them directly.\n\nChange-Id: I5bad754ef7b1bec6986c2b4293a38226c5a261af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8896740bbe2af6244796cc46a873b70048c636a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/8896740bbe2af6244796cc46a873b70048c636a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
