)]}'
{"id":"openstack%2Fsushy-tools~842385","triplet_id":"openstack%2Fsushy-tools~master~Ib7c3267f401355cf045954e71fe7297b484508be","project":"openstack/sushy-tools","branch":"master","topic":"armscsi","attention_set":{},"removed_from_attention_set":{"34740":{"account":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"},"last_update":"2022-06-27 01:05:45.000000000","reason":"\u003cGERRIT_ACCOUNT_34740\u003e replied on the change","reason_account":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"}}},"hashtags":[],"change_id":"Ib7c3267f401355cf045954e71fe7297b484508be","subject":"aarch64/arm64 libvirt VMs should default to SCSI controller","status":"MERGED","created":"2022-05-18 16:05:03.000000000","updated":"2022-06-27 03:15:55.000000000","submitted":"2022-06-27 03:14:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842385-armscsi","meta_rev_id":"429ccfbbc117b6729d71eabe4e375613b104ef2d","_number":842385,"virtual_id_number":842385,"owner":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-27 03:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":2,"date":"2022-05-18 22:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-01 07:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-01 07:02:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 17:13:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 22:45:18.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2022-06-01 07:02:16.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"8d4ce7f15f1033a32514719c5b6f97bb90e34095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"},"date":"2022-05-18 16:05:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ca784c0a2fe6533ba75038447130cba258b5a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 17:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f5035831e4419b94c44c3f49a012e2 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93aabecc98a54dcfbf754bb55c39106e : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5f350650f444f189bc06c533ce6e342 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7637c77e1e1549a8b14d8d58e32a0b2d : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f15ba0c0624eb2888bf63c58ab1ce7 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/940266353124466d818848fac784bc3c : SUCCESS in 3m 53s\n- sushy-tools-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/592cf44fe6264caebdbcdfc1f73b13c9 : SUCCESS in 52m 40s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9cf02e5e59e34adfb6f8160d9be64245 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ebb51bdbd5ba1994ac7b6f8ffc4d8c396cda786d","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-05-18 22:45:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03a3bcafb2f05536314ae6e7a8f21fd312dd1883","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-06-01 07:02:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f881bbf4a93b8feeb90c78ff641f56c5af8203a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 07:02:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"125497266c7029ce5dcdde0183601e5e97ff1769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 08:00:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad62e3892910470a922b31b3feeaf590 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6678bd46ab04e92a4a51a5b895a4894 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a84fa7e2d9684591adcfd5db7708428c : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29285bad4a174486aa6101578d439df4 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c81009968a604394b3756e1bc8dc2f49 : FAILURE in 3m 25s\n- sushy-tools-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/5ea1eb69973c4182aaeeffe4a42de380 : SUCCESS in 55m 20s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6b9f7d83a02a44198206129f6ccf91fd : SUCCESS in 50m 18s","accounts_in_message":[],"_revision_number":1},{"id":"285cd600054b4256acbec895484ee4cda73b1c05","author":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"},"date":"2022-06-24 18:13:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5015bfb8c5c0e16f2f04f9ec2177f45073f12dc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 19:45:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4157f4cb08104fe7a4107b384470b1de : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6584128bae654de699d1bbafb9da7e1b : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ed5c0201674917a48be679dbf551da : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46a34e7862594304a6c3b47ccb90dfee : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0092adb57207494fa551766c534a266e : SUCCESS in 5m 52s\n- sushy-tools-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/3cf0d542c1e147319cb1093fb4a455f0 : FAILURE in 1h 28m 42s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/48e38c4689e94576ad085941ce9c0678 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":1},{"id":"329fc2fa2fda5d5defbdb353b57257b32d68b70f","author":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"},"date":"2022-06-27 01:05:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"347f9ef547a1170b57a0cba521c268bcfcffab24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 02:07:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34f528da82fe45c7be4bef5ceec6bc7f : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e4da45b8546497bafd36085c078784f : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9c20a714f0142b086fa9af4f5f180c0 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e444b0da02b408cabf08bf1343659c4 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705915d3b0b846b29196fa3d82bc900b : SUCCESS in 6m 48s\n- sushy-tools-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/370d387837c444b7af0ef9d0ce8546ca : SUCCESS in 55m 13s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/445057071fe74b17bc5e4db22a803d51 : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":1},{"id":"95b77bd97633e83f0a59f51fe2c09c161bdc2cc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 02:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28feb28838dab487410aea52deb2eb7fcce5a2b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 03:14:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6af64177000437787526a2426e51859 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54190c20093496f904edf66a0d15b11 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0923a48093204d459cd987124f871555 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b081def178c463b963520d701c2d4c4 : SUCCESS in 5m 36s\n- sushy-tools-tempest-bios-redfish-netboot https://zuul.opendev.org/t/openstack/build/e12b4c3caa0f48d088ba8794dfdb2b50 : SUCCESS in 49m 07s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3ebde91fe91d4fad9ac2fd056f5e74a6 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b7292bbedac67a07eb50a2deeaef79bd57ca9d6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 03:14:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"429ccfbbc117b6729d71eabe4e375613b104ef2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 03:15:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6931cd392cf4372afef63d854f3cf12 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0acc3de080abd1efe6460d9255f065807b6fb81","revisions":{"c0acc3de080abd1efe6460d9255f065807b6fb81":{"kind":"REWORK","_number":1,"created":"2022-05-18 16:05:03.000000000","uploader":{"_account_id":34740,"name":"Jeff Young","email":"jeyoung@redhat.com","username":"jeyoung"},"ref":"refs/changes/85/842385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/85/842385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/842385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/842385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/85/842385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/85/842385/1"}}},"commit":{"parents":[{"commit":"b0e7380047ba127d2ac57992665af737bb2139d7","subject":"Merge \"Update jobs names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b0e7380047ba127d2ac57992665af737bb2139d7"}]}],"author":{"name":"Jeff Young","email":"jeyoung@redhat.com","date":"2022-05-18 14:01:06.000000000","tz":-240},"committer":{"name":"Jeff Young","email":"jeyoung@redhat.com","date":"2022-05-18 14:01:06.000000000","tz":-240},"subject":"aarch64/arm64 libvirt VMs should default to SCSI controller","message":"aarch64/arm64 libvirt VMs should default to SCSI controller\n\nThe current logic breaks when sushy-tools tries to mount virtual media to an arm64/aarch64 libvirt vm.\nThis is because by default it tries to connect and ide controller, and the only controller option for arm is scsi.\n\nThis patch ensures that aarch64 type vms will always get a scsi controller.\n\nStory: 2009962\nTask: 44940\nChange-Id: Ib7c3267f401355cf045954e71fe7297b484508be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c0acc3de080abd1efe6460d9255f065807b6fb81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c0acc3de080abd1efe6460d9255f065807b6fb81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[]}
