)]}'
{"id":"openstack%2Fsushy-tools~851819","triplet_id":"openstack%2Fsushy-tools~master~I53703f067fb41c27e42a107883ad2562c6be4edb","project":"openstack/sushy-tools","branch":"master","attention_set":{},"removed_from_attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2022-08-02 11:14:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53703f067fb41c27e42a107883ad2562c6be4edb","subject":"remove unicode from code","status":"MERGED","created":"2022-08-02 08:26:58.000000000","updated":"2022-08-02 11:15:08.000000000","submitted":"2022-08-02 11:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851819","meta_rev_id":"d751f5e70c3cb8a1b1ea7b40e0d46d5f53f883f7","_number":851819,"virtual_id_number":851819,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-02 11:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-08-02 10:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-08-02 10:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 09:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 10:06:07.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"5b781167118708074891fab5ca7c0ce1ed68f54c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-02 08:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2596a0675a763924109909cbcdacb2a603edb9c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 09:38:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21108d90aa6483ba5d8aec2c046f07d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9899c90bc0d64830b47a4a5d0c088945 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045f1cc70a0a4e68962bd47d7117f80e : SUCCESS in 4m 52s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/dd07708c7e624f8abbf979d60d642b5f : SUCCESS in 47m 58s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d59a5430a7d54f0585b1337963b5971e : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":1},{"id":"459bd504d9b96da6018c8063b0305526233e59c3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-02 10:06:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"941ceb146ab18895f2dbe18eac09da2286e3ba94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:06:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3feec40b5dfec85f96e742bf9692f66e9d8a472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:14:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/debaec5cc5a6489ea2b582c50b6180cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a5e0cb5edb4f87bc8e9bc6604895b6 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77ca29ddecc42e6b5307f0de9c5fee7 : SUCCESS in 7m 04s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/65cb73cc5fec445eba7c26399e90526b : SUCCESS in 37m 20s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/33d107e0c7a4479c9db02fd86eb31b55 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7d7fd2089c8ff27fa37f52d2ba6231941591634e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:14:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d751f5e70c3cb8a1b1ea7b40e0d46d5f53f883f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:15:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100900df30134e5ab9920d84cdf18a52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9808c4679034261a1b72e843d260acd : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d576943e056b2ca37e1fe0bbce1d1587705ac704","revisions":{"d576943e056b2ca37e1fe0bbce1d1587705ac704":{"kind":"REWORK","_number":1,"created":"2022-08-02 08:26:58.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/19/851819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/19/851819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/19/851819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/19/851819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/19/851819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/19/851819/1"}}},"commit":{"parents":[{"commit":"23362129ed399de0e7061b1e91d669b3f4a36b66","subject":"Merge \"Document SUSHY_EMULATOR_ALLOWED_INSTANCES\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/23362129ed399de0e7061b1e91d669b3f4a36b66"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-02 08:26:36.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-02 08:26:43.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: I53703f067fb41c27e42a107883ad2562c6be4edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d576943e056b2ca37e1fe0bbce1d1587705ac704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/d576943e056b2ca37e1fe0bbce1d1587705ac704"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
