)]}'
{"id":"openstack%2Fsushy-tools~874560","triplet_id":"openstack%2Fsushy-tools~master~If44be26e3fba70b89aec6ba88743d740276e7560","project":"openstack/sushy-tools","branch":"master","topic":"vmedia_ignore_file_not_found","attention_set":{},"removed_from_attention_set":{"34490":{"account":{"_account_id":34490,"name":"Michele Costa","display_name":"micosta","email":"micosta@redhat.com","username":"micosta"},"last_update":"2023-02-21 16:51:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If44be26e3fba70b89aec6ba88743d740276e7560","subject":"Ignore FileNotFoundError when ejecting virtual media image","status":"MERGED","created":"2023-02-21 11:52:10.000000000","updated":"2023-02-21 16:52:10.000000000","submitted":"2023-02-21 16:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874560-vmedia_ignore_file_not_found","meta_rev_id":"d65ec62fcd7c7cc5473e4baa8a014327d2112f9a","_number":874560,"virtual_id_number":874560,"owner":{"_account_id":34490,"name":"Michele Costa","display_name":"micosta","email":"micosta@redhat.com","username":"micosta"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 16:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-02-21 14:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-21 14:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2023-02-21 14:30:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 12:38:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 14:16:04.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2023-02-21 14:30:31.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"9e8d80d833cf7a4deb2b60556b7235097677dc24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34490,"name":"Michele Costa","display_name":"micosta","email":"micosta@redhat.com","username":"micosta"},"date":"2023-02-21 11:52:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71be047058c6e46903be05855d503edc7ea7efc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:38:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75e919da2bc14e10bf448c2299cfa95a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2076157291ef4d0397d55fe6fce7102a : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a59cc6cc57cb4d7285e93770b825867b : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2369e3c212ba42aa821171e5ff550737 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9512ae46504f41a6be541590c178e69c : SUCCESS in 5m 29s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/461f0a45530d4ff99c88a0fc108b9ff6 : SUCCESS in 37m 27s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/530630cd803d4cfcb852e044762b9b6c : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f4794d76f37f2393eddc82d93d33fa268ecfedaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34490,"name":"Michele Costa","display_name":"micosta","email":"micosta@redhat.com","username":"micosta"},"date":"2023-02-21 14:15:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83aaaafa7482a818c5747f396d42f0aa9e59097b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-02-21 14:16:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7935a24ee26853f123019249bce98f1110fdd988","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-02-21 14:30:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0e8001f580a5a287d547df0f02401af25153ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:41:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/334a173c30be47088d5569c17706f56b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888ae0f8d1da44d38c3e98a87d9a0431 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d199237f33c14337ab498e3b68709054 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9973a69b5a9045d2a432c9bb0ba53b82 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b01734f7a1f4eb09a1314471c93e368 : SUCCESS in 4m 17s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/59627705d9024cb38c491fa5de59f5ba : SUCCESS in 32m 53s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d8f02c8afe2b4454b175cfb072692f96 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":2},{"id":"7de9e69b33b32d9104728326896971cf1dd34fc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:42:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ed949414c1fbab98023b91967b4a4c8a304727c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:51:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1ca197d83a4ed1b7f0abc4dc7dfc3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797807cdaa2840698fd1659a549e6078 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfe676a41b31447a885c228467b75b77 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fde027e4d1a04d0b8ebdc7f3a0ee434a : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6493a9da62ab41dc9763358b703a10af : SUCCESS in 5m 50s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e6ae9e0a8d2b44919aa3869df739bccd : SUCCESS in 42m 03s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d6d3ebd10ae4477b9de7a84f6a203268 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f7a103c64225b8f7eb84ce71c56892a0370cdc09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:51:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d65ec62fcd7c7cc5473e4baa8a014327d2112f9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:52:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8d674f680a442e78028a1864bbafa73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d774e6b8c5849629d8fc7bc3c9b7ebb : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4357bf2347c8aa85fce305ff4be4a50d475a7bce","revisions":{"e621538cb33ded3fbcfa1e03f61fd511eaea6a25":{"kind":"REWORK","_number":1,"created":"2023-02-21 11:52:10.000000000","uploader":{"_account_id":34490,"name":"Michele Costa","display_name":"micosta","email":"micosta@redhat.com","username":"micosta"},"ref":"refs/changes/60/874560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/60/874560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/1"}}},"commit":{"parents":[{"commit":"7115db2f3adc6db673fa87830e15e22d812f5783","subject":"Merge \"Save previous hd boot device when setting Pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7115db2f3adc6db673fa87830e15e22d812f5783"}]}],"author":{"name":"Michele Costa","email":"micosta@redhat.com","date":"2023-02-21 11:50:03.000000000","tz":0},"committer":{"name":"Michele Costa","email":"micosta@redhat.com","date":"2023-02-21 11:52:09.000000000","tz":0},"subject":"Ignore FileNotFoundError when ejecting virtual media image","message":"Ignore FileNotFoundError when ejecting virtual media image\n\nAs these files are in /tmp the can be garbage collected\ncausing a FileNotFoundError. As it is trying to remove\nthe file anyway there is no reason to propogate this\nerror as the end result is the same.\n\nChange-Id: If44be26e3fba70b89aec6ba88743d740276e7560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e621538cb33ded3fbcfa1e03f61fd511eaea6a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/e621538cb33ded3fbcfa1e03f61fd511eaea6a25"}]},"branch":"refs/heads/master"},"4357bf2347c8aa85fce305ff4be4a50d475a7bce":{"kind":"REWORK","_number":2,"created":"2023-02-21 14:15:19.000000000","uploader":{"_account_id":34490,"name":"Michele Costa","display_name":"micosta","email":"micosta@redhat.com","username":"micosta"},"ref":"refs/changes/60/874560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/60/874560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/60/874560/2"}}},"commit":{"parents":[{"commit":"7115db2f3adc6db673fa87830e15e22d812f5783","subject":"Merge \"Save previous hd boot device when setting Pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7115db2f3adc6db673fa87830e15e22d812f5783"}]}],"author":{"name":"Michele Costa","email":"micosta@redhat.com","date":"2023-02-21 11:50:03.000000000","tz":0},"committer":{"name":"Michele Costa","email":"micosta@redhat.com","date":"2023-02-21 14:14:33.000000000","tz":0},"subject":"Ignore FileNotFoundError when ejecting virtual media image","message":"Ignore FileNotFoundError when ejecting virtual media image\n\nAs these files are in /tmp the can be garbage collected\ncausing a FileNotFoundError. As it is trying to remove\nthe file anyway there is no reason to propogate this\nerror as the end result is the same.\n\nChange-Id: If44be26e3fba70b89aec6ba88743d740276e7560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4357bf2347c8aa85fce305ff4be4a50d475a7bce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4357bf2347c8aa85fce305ff4be4a50d475a7bce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[]}
