)]}'
{"id":"openstack%2Fsushy-tools~912113","triplet_id":"openstack%2Fsushy-tools~master~I5d46ee94d22e7c0913305070e366e4b06226dc5b","project":"openstack/sushy-tools","branch":"master","topic":"image_query_issue","attention_set":{"35233":{"account":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"last_update":"2024-05-28 16:03:20.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2024-06-06 14:01:25.000000000","reason":"\u003cGERRIT_ACCOUNT_36770\u003e replied on the change","reason_account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}}},"hashtags":[],"change_id":"I5d46ee94d22e7c0913305070e366e4b06226dc5b","subject":"improve volume support in nova driver","status":"NEW","created":"2024-03-08 13:23:23.000000000","updated":"2024-08-21 08:10:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":29,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b4753e949e2f5240fca12f4da9ef951366abca26","_number":912113,"virtual_id_number":912113,"owner":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-28 17:05:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"disliked":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},{"value":1,"date":"2024-06-06 14:01:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-08-21 08:10:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-08 14:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 18:52:41.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2024-08-21 08:10:53.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"091be063814f5e6e3526b0518528cf337e52b6e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-08 13:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8151a26cfa588043cca2bb71ef5f913af47fecb5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-08 13:23:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"23b51f3cbd6a6aeb77760a5e6cc716aad7a9a7bf","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-08 13:24:07.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3bf91f22adf7fe0327abc74b60b26b308a6bbd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 14:19:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ed342647a4944b5a5591f15a6fd29d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6ce770a39d4b7fa416b642c23d03a8 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf58ae84dfe47a894ebc0a153e3fa7a : FAILURE in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5038cb8438a44e56bc24b701da62eb86 : FAILURE in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d30e88ddf242f3bbb5c4796e21e71f : SUCCESS in 5m 04s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/711c14edeec9497cb87c0ae9f9981cd8 : SUCCESS in 30m 50s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2106dd3eb39340f0b26e594b73529438 : SUCCESS in 53m 45s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/21ec9c7843764e13b2ffe32bb8a6f984 : FAILURE in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba25779619317fd9529459d2acc1f60d734673ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-11 09:28:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77261cc3f9472e76e39b35300b777a8f028d54cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 10:45:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78eb25f36a584cb8abe5e75988f53e78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb468c2c72904d81a081338b23f82695 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86ce701705f844b9a179d59a64acbac3 : FAILURE in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54c064f74ee24fb88dc324e426d33841 : FAILURE in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5afc643269a341f497abdfeb6e66a286 : SUCCESS in 6m 17s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7a06290bb1be4419821adc718c1fbc47 : SUCCESS in 45m 28s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ea6059aa30104e90848c0cf2aa2c20bd : SUCCESS in 1h 04m 08s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/cb67be207e084650be504bee62e6d2c8 : FAILURE in 2m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e2fabec86bf2f24f7d1e41d378182e2f487015a","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-11 17:50:25.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"a091268e99fd1ae6cbd053146244ba6ec3bbc989","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-13 10:34:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c8cf41f8c1607b36323c3a41790dc75d9d375e0e","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-13 10:36:49.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"374b7378bf831b70b9668fef9fa4182cab0f6d6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-13 11:08:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4f3b5ee875203693dde0ec83b27bb88af31a33f3","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-03-13 11:09:37.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"cddbe444f68d4fcb1a50179bbde13604be665530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:18:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead60b7a166843bea66b3dc021f466ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e6dc24c3de4aadbabd88048648f935 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f93ad9ccdbc44e3b36e485df7f524ef : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fda4ee77c48f49b38c2b90ad599f6fc8 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6448dd2d57584296a09edd791423c643 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4718c9ae6dcf433ea00373ed785b2a7e : SUCCESS in 3m 01s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5eca2d28cd9e464eab6ced4ca21cbae9 : SUCCESS in 44m 23s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/400b2df841704d25a2b141c9c26f9eac : SUCCESS in 1h 03m 00s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/1a26131da7ff4045b2434609f123023a : FAILURE in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d691b95e46ba82aa0c191eeb740e52be246a5b7","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-03-13 18:52:41.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0798fa268cbd71523793d312cfa9d7e279024e10","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-24 11:58:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"89f6b7360adcb97a604791fb7f034852e7a6e19a","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-24 12:00:20.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"66d6f092f3b7bdfa5b49807b821818a0ab8ece55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:49:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f2c63004b7345f5b7c7e1e12d44af61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77168a4baae041799c976b832257fb6c : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce22d26db4a8437587903da21d038587 : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93350532119a446f987f2f3a81ccc059 : SUCCESS in 8m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/188642f755ea46aa84ce0ba0ba9e5634 : FAILURE in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af508dfcf254c8f985faaaec85426bb : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d065a9898324573a64035ebd9f9948a : SUCCESS in 2m 29s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0e9ac7e762eb43fd8657ed23d148b833 : SUCCESS in 43m 48s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0aeefa46401844f4843bf6fbc53aa4f2 : SUCCESS in 49m 13s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/841a95511c294c5b800fef1469c1df80 : FAILURE in 2m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"300c484fe4916e85582318261db973e5ca61d597","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-28 15:38:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"f0c87a2def2e500d54618da7e848f15201ff7ce1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-28 16:02:51.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"a6d73632a752efc47e0e411c666128b1c546c4a9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-28 16:03:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"e7ded982934cb1e44d58eb2e83cc116c2c52fcaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-28 16:05:01.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"6c1a4203aeeffd18584468bf20e8104080619451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-28 16:06:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c53c3e70d87ffa99b225373f91f65c1d17a8a800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"date":"2024-05-28 16:08:33.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"c0031f4b95d74cf5df8f8c1427c2b0ad4b08826d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:05:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55cefc8929f64aa6986c70c3ba4b23ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19244a8fe6e433daa155ee490b82634 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eece5d1341a47389c15f56a31f9521f : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f640702283024b909e21abcd6a3532ad : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99ad23a435e24d84b669dbb6a275844d : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41549d9c9b964108952b4ffa9f710aa4 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23ae97a08fe348ceab2a19a51f6eb3b6 : SUCCESS in 2m 36s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/39ba4acc4f614d67966b355ef3c792d2 : SUCCESS in 45m 56s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e7cff6855dbb478db2968fa5c3c3ec9f : SUCCESS in 39m 15s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/4a5a3478f8c7406491783578ee419002 : FAILURE in 2m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7b03d719099a7d021723fc33c81a1a56227fac27","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-06-06 14:01:25.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"475f0d2524da6a3b4cc6053dbc5ca935bf7a07ac","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-06-06 14:03:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b4753e949e2f5240fca12f4da9ef951366abca26","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-08-21 08:10:53.000000000","message":"Patch Set 10: Code-Review-1\n\n(25 comments)\n\nThe direction looks good, some comments inline (mostly around error handling).","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f16ed38fe87eef967983b92e77aab55a5565785c","revisions":{"f3ac8ae0553dafefb6a6f12f5f670b89719df2ca":{"kind":"REWORK","_number":1,"created":"2024-03-08 13:23:23.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/1"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-03-08 13:23:20.000000000","tz":120},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f3ac8ae0553dafefb6a6f12f5f670b89719df2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f3ac8ae0553dafefb6a6f12f5f670b89719df2ca"}]},"branch":"refs/heads/master"},"de8167f2b4f979ca6de15eae65aeec888bf09cc5":{"kind":"REWORK","_number":2,"created":"2024-03-11 09:28:23.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/2"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-03-11 09:28:11.000000000","tz":120},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/de8167f2b4f979ca6de15eae65aeec888bf09cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/de8167f2b4f979ca6de15eae65aeec888bf09cc5"}]},"branch":"refs/heads/master"},"cec0a2518ff9da39d825ccb81dc20b40f404bd45":{"kind":"REWORK","_number":3,"created":"2024-03-13 10:34:20.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/3"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-03-13 10:33:36.000000000","tz":120},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/cec0a2518ff9da39d825ccb81dc20b40f404bd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/cec0a2518ff9da39d825ccb81dc20b40f404bd45"}]},"branch":"refs/heads/master"},"4d88da5cd9484de95c5bfc7c2b78312154d6da2a":{"kind":"REWORK","_number":4,"created":"2024-03-13 11:08:41.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/4"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-03-13 11:08:35.000000000","tz":120},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d88da5cd9484de95c5bfc7c2b78312154d6da2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4d88da5cd9484de95c5bfc7c2b78312154d6da2a"}]},"branch":"refs/heads/master"},"c6643a476701dbc041e197117aca9fa540854f50":{"kind":"REWORK","_number":5,"created":"2024-05-24 11:58:45.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/5"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-05-24 11:58:16.000000000","tz":180},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c6643a476701dbc041e197117aca9fa540854f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/c6643a476701dbc041e197117aca9fa540854f50"}]},"branch":"refs/heads/master"},"73814019b8ff66c3172a2b309912f2aaaa2d4597":{"kind":"REWORK","_number":6,"created":"2024-05-28 15:38:41.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/6"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-05-28 12:55:02.000000000","tz":180},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/73814019b8ff66c3172a2b309912f2aaaa2d4597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/73814019b8ff66c3172a2b309912f2aaaa2d4597"}]},"branch":"refs/heads/master"},"330d5ca2ff072e734f98d192b87dd3f5a1effe1a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-05-28 16:02:51.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/7"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-05-28 16:02:08.000000000","tz":180},"subject":"improve volume boot mode support","message":"improve volume boot mode support\n\nThis commit introduces the following changes:\n - Support for VMs that have been boot from OpenStack volumes\n - Support for VMs that have been booted from OpenStack volume snapshots\n - New Nova driver functionality to be able to traverse volume chains\n   in order to find the bootable root image of given chain.\n - Customizable volume chain traversal depth via configuration variable\n   SUSHY_MAX_VOLUME_DEPTH (default 15)\n\nCloses-Bug: #2067405\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/330d5ca2ff072e734f98d192b87dd3f5a1effe1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/330d5ca2ff072e734f98d192b87dd3f5a1effe1a"}]},"branch":"refs/heads/master"},"5186f5146939a17d0d4024f3701841d633f94d89":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-05-28 16:05:01.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/8"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-05-28 16:04:19.000000000","tz":180},"subject":"improve volume support in nova driver","message":"improve volume support in nova driver\n\nThis commit introduces the following changes:\n - Support for VMs that have been boot from OpenStack volumes\n - Support for VMs that have been booted from OpenStack volume snapshots\n - New Nova driver functionality to be able to traverse volume chains\n   in order to find the bootable root image of given chain.\n - Customizable volume chain traversal depth via configuration variable\n   SUSHY_MAX_VOLUME_DEPTH (default 15)\n\nCloses-Bug: #2067405\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5186f5146939a17d0d4024f3701841d633f94d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/5186f5146939a17d0d4024f3701841d633f94d89"}]},"branch":"refs/heads/master"},"b262790ec9e4fdc03fd9c32d743b661f0f24908f":{"kind":"REWORK","_number":9,"created":"2024-05-28 16:06:42.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/9"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-05-28 16:06:36.000000000","tz":180},"subject":"improve volume support in nova driver","message":"improve volume support in nova driver\n\nThis commit introduces the following changes:\n - Support for VMs that have been boot from OpenStack volumes\n - Support for VMs that have been booted from OpenStack volume snapshots\n - New Nova driver functionality to be able to traverse volume chains\n   in order to find the bootable root image of given chain.\n - Customizable volume chain traversal depth via configuration variable\n   SUSHY_MAX_VOLUME_DEPTH (default 15)\n\nCloses-Bug: #2067405\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b262790ec9e4fdc03fd9c32d743b661f0f24908f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b262790ec9e4fdc03fd9c32d743b661f0f24908f"}]},"branch":"refs/heads/master"},"f16ed38fe87eef967983b92e77aab55a5565785c":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2024-05-28 16:08:33.000000000","uploader":{"_account_id":35233,"name":"Adam Rozman","email":"adam.rozman@est.tech","username":"rozzix"},"ref":"refs/changes/13/912113/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/13/912113/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/13/912113/10"}}},"commit":{"parents":[{"commit":"382a548e316136973bf337563856257a7a59c0d2","subject":"Handle nova instance having no image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/382a548e316136973bf337563856257a7a59c0d2"}]}],"author":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-02-27 08:28:37.000000000","tz":120},"committer":{"name":"Adam Rozman","email":"adam.rozman@est.tech","date":"2024-05-28 16:08:18.000000000","tz":180},"subject":"improve volume support in nova driver","message":"improve volume support in nova driver\n\nThis commit introduces the following changes:\n - Support for VMs that have been booted from OpenStack volumes\n - Support for VMs that have been booted from OpenStack volume snapshots\n - New Nova driver functionality to be able to traverse volume chains\n   in order to find the bootable root image of given chain.\n - Customizable volume chain traversal depth via configuration variable\n   SUSHY_MAX_VOLUME_DEPTH (default 15)\n\nCloses-Bug: #2067405\nChange-Id: I5d46ee94d22e7c0913305070e366e4b06226dc5b\nSigned-off-by: Adam Rozman \u003cadam.rozman@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f16ed38fe87eef967983b92e77aab55a5565785c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f16ed38fe87eef967983b92e77aab55a5565785c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
