)]}'
{"id":"openstack%2Fsushy-tools~942212","triplet_id":"openstack%2Fsushy-tools~master~Ibceaef4cc1888bd9101e8054289ea7beddcfcf33","project":"openstack/sushy-tools","branch":"master","topic":"os_vmedia","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2025-02-20 00:03:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibceaef4cc1888bd9101e8054289ea7beddcfcf33","subject":"Openstack - get/set_boot_image support volume boot","status":"MERGED","created":"2025-02-19 15:37:12.000000000","updated":"2025-02-20 00:04:06.000000000","submitted":"2025-02-20 00:03:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942212-os_vmedia","meta_rev_id":"69ca8061c2f80d54f543ca0c38d3c5d3c7fb182b","_number":942212,"virtual_id_number":942212,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 00:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-02-19 20:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-02-19 19:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-02-19 20:00:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 16:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 19:29:54.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2025-02-19 20:00:29.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"5519161162628f67854484682ee6ee5816e2fc05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-02-19 15:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1d39ce4af3654e9a1533b83316c993b2b5d8d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 16:27:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbba3fddbd7a483ea25eef7ba9571411\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83aa69507134165957c9315613fe276 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44564599a0e54b9899aa1d2a9a694366 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3cf1850b46004de8a751072f9495884b : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8471bb74dc842a6a4cdbff87c0d64d5 : SUCCESS in 4m 37s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a4db489bc16a48daae3da2ce7c86a062 : SUCCESS in 37m 33s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/de6ffb80093942f686b53ff2ad70659a : SUCCESS in 44m 18s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/b76a0e3b3d134bebafe3ead13a01eec8 : FAILURE in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b5eee9370026e1a987472cf0f42d3a8bfab267a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-02-19 19:28:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"59536122ac17bcffdfc90bf423ab2a8ffaf7e498","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2025-02-19 19:29:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8406c3fe24d982cfa8a713fb3d981d5e3608675a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2025-02-19 19:55:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1002ef9478fa8e23a33b3d405ef1e3695df34fae","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-02-19 20:00:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2a5d38857a64404c89f899061159d7b8adc48d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 20:31:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff54de083e37423eb66b41067f049b02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f19e8bf6e9d48b2bf6d658eb61ca92f : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9880ed3141274693930b68897a0e818e : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a994558b11d246d0bcbcac9c6f64a2cc : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df609b309094b859e492895ce6bbb63 : SUCCESS in 4m 56s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3db02e80e8a54a808d3dc4e2b27ef071 : SUCCESS in 24m 48s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5ec0961a3041484e9dee29d47074d3f1 : SUCCESS in 35m 28s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/636f2c4de5a440dea7a13323c0f85c93 : FAILURE in 1m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c238aeb68f45d3484232fd71c894804f0a57c9a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 23:01:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1492635cdb4d64a131e97baefa201231ca273d13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 00:03:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6c1fd196a2144dcbeffa760d31f6ffa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0fb9a0b0f341b393b651e9caf6167f : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecdeb29164ab4283bdc84863e67f1b76 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68ef4e55f5364889a4e8b20a8631eaa3 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da1083dcc2e42e7bdac87519dd8d32c : SUCCESS in 4m 10s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/770470c7115046c5a1e630790e25414f : SUCCESS in 30m 58s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7769f137d21d4bbd8f152cc7ba38391a : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":3},{"id":"2b2f32f2c337192bf87b8d20fc2bc663f3b19865","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 00:03:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"69ca8061c2f80d54f543ca0c38d3c5d3c7fb182b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 00:04:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d5a430cd2c46fc844af663081f87b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e504ce491847a8a2d2799c612a7e54 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7dd7c13bbaf912117cb6fd584eb9e80ded5ea705","revisions":{"f9309d23a67dbb02ae05f91abe5b57a5ab0bb9a6":{"kind":"REWORK","_number":1,"created":"2025-02-19 15:37:12.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/12/942212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/12/942212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/1"}}},"commit":{"parents":[{"commit":"4c7a6ce2c0f40ea7a486c83f26dc8d304e036fe7","subject":"Openstack - get_boot_mode support volume boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/4c7a6ce2c0f40ea7a486c83f26dc8d304e036fe7"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-02-11 23:21:38.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-02-19 15:24:55.000000000","tz":60},"subject":"Openstack - get/set_boot_image support volume boot","message":"Openstack - get/set_boot_image support volume boot\n\ninstance.image.id is always None for boot from volume instance, when\nthis is the case get the volumes attached to the instance and attempt to\nget the image id of the volume metadata.\n\nChange-Id: Ibceaef4cc1888bd9101e8054289ea7beddcfcf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f9309d23a67dbb02ae05f91abe5b57a5ab0bb9a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/f9309d23a67dbb02ae05f91abe5b57a5ab0bb9a6"}]},"branch":"refs/heads/master"},"9663343c743a65630a4d167b47a0e0204aca639a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-19 19:28:54.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/12/942212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/12/942212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/2"}}},"commit":{"parents":[{"commit":"85713b57ff801732d8abeb9a644bb72b2717a740","subject":"Openstack - get_boot_mode support volume boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/85713b57ff801732d8abeb9a644bb72b2717a740"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-02-11 23:21:38.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-02-19 19:27:56.000000000","tz":60},"subject":"Openstack - get/set_boot_image support volume boot","message":"Openstack - get/set_boot_image support volume boot\n\ninstance.image.id is always None for boot from volume instance, when\nthis is the case get the volumes attached to the instance and attempt to\nget the image id of the volume metadata.\n\nChange-Id: Ibceaef4cc1888bd9101e8054289ea7beddcfcf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/9663343c743a65630a4d167b47a0e0204aca639a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/9663343c743a65630a4d167b47a0e0204aca639a"}]},"branch":"refs/heads/master"},"7dd7c13bbaf912117cb6fd584eb9e80ded5ea705":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-19 19:55:03.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/12/942212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/12/942212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/12/942212/3"}}},"commit":{"parents":[{"commit":"474443e7d3585ff29fc04ecd36d753a1a9228eea","subject":"Openstack - get_boot_mode support volume boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/474443e7d3585ff29fc04ecd36d753a1a9228eea"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-02-11 23:21:38.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2025-02-19 19:54:22.000000000","tz":60},"subject":"Openstack - get/set_boot_image support volume boot","message":"Openstack - get/set_boot_image support volume boot\n\ninstance.image.id is always None for boot from volume instance, when\nthis is the case get the volumes attached to the instance and attempt to\nget the image id of the volume metadata.\n\nChange-Id: Ibceaef4cc1888bd9101e8054289ea7beddcfcf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7dd7c13bbaf912117cb6fd584eb9e80ded5ea705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/7dd7c13bbaf912117cb6fd584eb9e80ded5ea705"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
