)]}'
{"id":"openstack%2Fsushy-tools~972222","triplet_id":"openstack%2Fsushy-tools~master~Ie3a777848527200ceb8d85ffbcea6660f1ea3384","project":"openstack/sushy-tools","branch":"master","topic":"conn-leak","attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-07-08 17:02:26.000000000","reason":"Vote got outdated and was removed: Workflow+1"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2026-07-08 17:54:36.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie3a777848527200ceb8d85ffbcea6660f1ea3384","subject":"memoize: do not leak open connections","status":"NEW","created":"2026-01-05 17:11:51.000000000","updated":"2026-07-08 17:54:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"85e466052e4ca90a90e87f1ab87429790af7db3f","_number":972222,"virtual_id_number":972222,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-08 17:54:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":-1,"date":"2026-07-08 17:02:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2026-07-08 17:02:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-07-08 17:02:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-08 17:02:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":-1,"date":"2026-07-08 17:02:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 17:12:46.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-01-05 18:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 18:11:08.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2026-01-06 22:04:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-07 09:22:28.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"fc612369be782f9c9cbe28f425bca631857ed40e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-01-05 17:11:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c3febd56a2e9e4d4ec95aca318570fae481ff9d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-01-05 17:12:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd02f062394cfdcdccafe58c1f77fecabcc4e22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:12:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/443aa71d5aaf4cb9800c73b3d2478193\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354de41182b840e2950405486596483b : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5425a0df225c4c19979fa60ee83f70dd : FAILURE in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b09bf15a18a04d6fac0cef8767733f30 : FAILURE in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59e5df5f60644fea6ee387e3c5be627 : SUCCESS in 6m 36s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d2b8c01208ae474a923d1f2a4bd1c0ba : SUCCESS in 46m 46s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f1066e2c6f824236ae584ab1716ffb34 : SUCCESS in 48m 15s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/8c41cc90240148c38f47a5c99e6a793f : SUCCESS in 3m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dbc15ca30fb6f9f6a95b9b00d2911a4a1e3b966","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-01-06 18:11:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8db94597e3adbc93b529c22b2a16170813e98b6a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-06 22:04:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dadf6486ed12c321f34a8f0f6a2529891f7220bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 22:42:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e0b2da564054fdfbd06e373cbeef781\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7957569e5be244a3bd0c31118de20be2 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa480d55c9e449a9adf861cbde13cc59 : FAILURE in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/295787d32534428dab47530c7abf01be : FAILURE in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac59a01e3a34b2ab7915bf33ea46fd5 : SUCCESS in 7m 42s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/74fa9a5edf3f4739b4baa4ee1da92f15 : SUCCESS in 19m 39s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2a76b875b00940099b6c238040431e63 : SUCCESS in 35m 18s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/9887009d52ca43cba58c048c8e0d564e : SUCCESS in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78d209878530f3d5fb408ecdf2fd70e8034d15a7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-01-07 09:22:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"371d3bb433692bc8b8aa4b1c1d9b226e0c367272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:11:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34a9805d6147443bbd69b77bbe7a7797\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c35b598efbc4fc38b0af82d3b5c719d : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8beb74b874e2430083184b038bd96400 : FAILURE in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a7e0bccfaac4e7ba391fcf80acff6da : FAILURE in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbba6efaeee14e98ae3e3ad2ac9a6256 : SUCCESS in 6m 01s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/17d2fc1eeb144997b96431013e75f84e : SUCCESS in 43m 48s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/43d92a9d2dfe4d16ad1dd0f814e0e672 : SUCCESS in 34m 03s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/c0f63ff8aae544888b2f53e07266734d : SUCCESS in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c684fed4cdbe329e27bce6912e8109c1b4821840","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-01-07 13:45:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a76e7bb64991f31da3f04a57c409cc2493dcfff2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-07-08 17:02:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8099a26059c8488135df8b51051faa82596d2093","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-07-08 17:02:40.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85e466052e4ca90a90e87f1ab87429790af7db3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 17:54:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/305a4b7d27704f3db4db35ca3e675862\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3c01abfa0b49cb99cba4847c3bd455 : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae8fb820d2044674adbaf96fb6914fbf : FAILURE in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67953e9bf71d4332af8fb44a23698ced : FAILURE in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aedbe06fe34b43eca5ecf86e48f2f978 : FAILURE in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09807216fdac4b3c956b011168278666 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b353dc3c9da47e7b621af641d8b82ed : SUCCESS in 4m 06s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/668f9b8645bc4929a608909b594c5846 : SUCCESS in 40m 07s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f339a28ae3dd484c8bdfb5eb3d9ae2d6 : SUCCESS in 46m 03s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/a4d143c0a6e04188bcd9e6a078a06e72 : SUCCESS in 2m 39s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c96018e4d48df1258721ce2c8a26334bb08fa08","revisions":{"6816bdb6b6a755c2386c447a6e0f4e9216453b5e":{"kind":"REWORK","_number":1,"created":"2026-01-05 17:11:51.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/22/972222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/22/972222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/1"}}},"commit":{"parents":[{"commit":"fe44ca776291c1f78b44adbd9c4801f5d237fba2","subject":"Merge \"Drop redundant variable definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/fe44ca776291c1f78b44adbd9c4801f5d237fba2"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-01-05 17:10:56.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-01-05 17:10:56.000000000","tz":60},"subject":"memoize: do not leak open connections","message":"memoize: do not leak open connections\n\nContrary to what the code expects, `with connection` does not cause the\nconnection to be closed in the end, leaking it.\n\nChange-Id: Ie3a777848527200ceb8d85ffbcea6660f1ea3384\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6816bdb6b6a755c2386c447a6e0f4e9216453b5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/6816bdb6b6a755c2386c447a6e0f4e9216453b5e"}]},"branch":"refs/heads/master"},"2c96018e4d48df1258721ce2c8a26334bb08fa08":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-08 17:02:26.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/22/972222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/22/972222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/22/972222/2"}}},"commit":{"parents":[{"commit":"0e88320d52aa9a5664dc1d11c8dd5e4947c3978a","subject":"Add optional ServiceRoot Vendor via SUSHY_EMULATOR_VENDOR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/0e88320d52aa9a5664dc1d11c8dd5e4947c3978a"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-01-05 17:10:56.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2026-07-08 17:02:13.000000000","tz":120},"subject":"memoize: do not leak open connections","message":"memoize: do not leak open connections\n\nContrary to what the code expects, `with connection` does not cause the\nconnection to be closed in the end, leaking it.\n\nChange-Id: Ie3a777848527200ceb8d85ffbcea6660f1ea3384\nSigned-off-by: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2c96018e4d48df1258721ce2c8a26334bb08fa08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2c96018e4d48df1258721ce2c8a26334bb08fa08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
