)]}'
{"id":"openstack%2Fsushy-tools~987486","triplet_id":"openstack%2Fsushy-tools~master~I6ac9bfdfa0bbb8992bfbcda7cce22ab3bca08d94","project":"openstack/sushy-tools","branch":"master","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2026-05-11 20:07:31.000000000","reason":"\u003cGERRIT_ACCOUNT_36770\u003e replied on the change","reason_account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},"39114":{"account":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"last_update":"2026-05-11 21:07:51.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2026-05-11 21:07:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ac9bfdfa0bbb8992bfbcda7cce22ab3bca08d94","subject":"Add --storage-pool CLI option","status":"MERGED","created":"2026-05-06 11:02:49.000000000","updated":"2026-05-11 21:08:50.000000000","submitted":"2026-05-11 21:07:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987486","meta_rev_id":"6354e54f22e7eeb78fef657a658a832be7f7796d","_number":987486,"virtual_id_number":987486,"owner":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 21:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2026-05-11 20:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2026-05-11 19:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":1,"date":"2026-05-11 20:07:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 11:58:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 13:14:32.000000000","updated_by":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-05-06 13:14:32.000000000","updated_by":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2026-05-11 19:52:45.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"8b2cbed127b0f3936200dec88ba30e6beb0d1153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"date":"2026-05-06 11:02:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04b20cd023eaa95ff97de613c293b6d7405ed56e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:58:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8329cd0674594d4a90f7ac46ff9e957f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6798fabb6b04ad3a3c3de2834061318 : FAILURE in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c36c219472f54b7c970971a3a296891d : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b58be0ee9bd542499ca788a8c8098917 : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71fa8a86bb5b4d099bf97f930a403bac : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c05d2675b61c41b9a855acf3fe2e9af7 : SUCCESS in 4m 11s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/10b644a8a7f34a7f977cfeb1d8d4fe06 : SUCCESS in 45m 56s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9f7830c774c24e49b072e1af209b8f06 : SUCCESS in 47m 44s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/31e38ec773f94b40a398e5e132fae4a1 : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78c123a31c4c44560283a9fb05a869823612c489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"date":"2026-05-06 12:54:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f2074d6a5b3d0dcb06cc6c4e1294075c48699e81","author":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"date":"2026-05-06 13:10:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc0b97b5fa919139f796a596a4841e78502c2ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:24:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d2b95f6c7143b4907b8e858e6a3be5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339eb2d261934971b5697c9e007eac1d : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e280cee9b94b4e04bbb69dbef274a735 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03005281bb754bff932e8a889a8fb633 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/412993a3ddde4d84a1c0eda0d2ed052a : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8406db099d59435881c6c4ee3d560c56 : SUCCESS in 6m 37s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f471a0842894443f8183b8ad74b02b97 : SUCCESS in 18m 45s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c7c4a8d2c3c042978efb212529baea21 : SUCCESS in 27m 14s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/3888dc31fc8e4604a542580400f9dc7e : SUCCESS in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08f973740ec792c22a246390195d73ad5db85646","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-11 19:52:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3d8c572bc13cf05f0f1ef320cbabb0544a756b14","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-05-11 20:07:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e35070a6631c43960fe19ffcb5fd41e143f1974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 20:08:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a35a8ea898cf19530c489f1d981a71ed3ea15cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 21:07:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a12b720b61a1484eb2d5d032acf354e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13f49337bd44d0d90477b5e6c424d1c : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10a5991b91f548a2a5f93175d67c6557 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ba280d7a4fc44169ec9c2cc888892b2 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a540fc231d346e498c34c68920db410 : SUCCESS in 6m 20s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f63b2542bb3340fc9cf2618e7bd48efd : SUCCESS in 22m 53s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9cde05eadd1b47119601e12df9cd1172 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7338401334fa4684e8d4e94669a5b249938ffce7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 21:07:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6354e54f22e7eeb78fef657a658a832be7f7796d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 21:08:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdfd31f93da4e73a6b8ba2ebc01c26d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/586cb2642eef4518b69700d438502176 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"621b2ab293aa4c0ad9e1685b2d06356ee023f912","revisions":{"36171fdf4276ec333151e39bcec28221fc40cc05":{"kind":"REWORK","_number":1,"created":"2026-05-06 11:02:49.000000000","uploader":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"ref":"refs/changes/86/987486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/86/987486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/1"}}},"commit":{"parents":[{"commit":"2b51921cc7cf248f14ae7e45d53f379643492cec","subject":"Verify power state after rescue operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2b51921cc7cf248f14ae7e45d53f379643492cec"}]}],"author":{"name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","date":"2026-05-06 10:58:05.000000000","tz":180},"committer":{"name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","date":"2026-05-06 11:02:37.000000000","tz":180},"subject":"Add --storage-pool CLI option","message":"Add --storage-pool CLI option\n\nAdd a new --storage-pool CLI argument to select the libvirt storage\npool used for volume creation. When provided, the value is stored in\nthe SUSHY_EMULATOR_STORAGE_POOL configuration key.\n\nChange-Id: I6ac9bfdfa0bbb8992bfbcda7cce22ab3bca08d94\nSigned-off-by: Tero Kauppinen \u003ctero.kauppinen@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/36171fdf4276ec333151e39bcec28221fc40cc05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/36171fdf4276ec333151e39bcec28221fc40cc05"}]},"branch":"refs/heads/master"},"621b2ab293aa4c0ad9e1685b2d06356ee023f912":{"kind":"REWORK","_number":2,"created":"2026-05-06 12:54:53.000000000","uploader":{"_account_id":39114,"name":"Tero Kauppinen","display_name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","username":"tero-kauppinen"},"ref":"refs/changes/86/987486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/86/987486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/86/987486/2"}}},"commit":{"parents":[{"commit":"2b51921cc7cf248f14ae7e45d53f379643492cec","subject":"Verify power state after rescue operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/2b51921cc7cf248f14ae7e45d53f379643492cec"}]}],"author":{"name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","date":"2026-05-06 10:58:05.000000000","tz":180},"committer":{"name":"Tero Kauppinen","email":"tero.kauppinen@est.tech","date":"2026-05-06 12:53:36.000000000","tz":180},"subject":"Add --storage-pool CLI option","message":"Add --storage-pool CLI option\n\nAdd a new --storage-pool CLI argument to select the libvirt storage\npool used for volume creation. When provided, the value is stored in\nthe SUSHY_EMULATOR_STORAGE_POOL configuration key.\n\nChange-Id: I6ac9bfdfa0bbb8992bfbcda7cce22ab3bca08d94\nSigned-off-by: Tero Kauppinen \u003ctero.kauppinen@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/621b2ab293aa4c0ad9e1685b2d06356ee023f912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/621b2ab293aa4c0ad9e1685b2d06356ee023f912"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
