)]}'
{"id":"openstack%2Fsushy-tools~991906","triplet_id":"openstack%2Fsushy-tools~master~I10885f643ae8642311600fe1c4839adb8f1f8c93","project":"openstack/sushy-tools","branch":"master","attention_set":{},"removed_from_attention_set":{"37442":{"account":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"last_update":"2026-09-22 18:52:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I10885f643ae8642311600fe1c4839adb8f1f8c93","subject":"Add ComputerSystem SerialNumber derived from domain UUID","status":"MERGED","created":"2026-06-05 17:00:36.000000000","updated":"2026-09-22 18:55:36.000000000","submitted":"2026-09-22 18:52:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991906","meta_rev_id":"1a0e96b8b1714be1190257128605a95b8fedfe4b","_number":991906,"virtual_id_number":991906,"owner":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 18:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2026-09-22 18:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-29 15:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2026-09-22 18:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-05 17:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-29 15:07:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2026-09-22 18:00:40.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"ae6e59c7a6611d18db6d0ba4f9f801aef2a3280c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2026-06-05 17:00:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a005fb5269d91d8ac30c84ae1d520e769aa2e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 17:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2c5a40cf304f619e9d1900b093ccac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8270f3ea47004676b5ce392bede94f4e : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a74045c3db5466cacad2805019e51e5 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/862c3c1b6d0245088b6811a029b3eaf0 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be214f71bc5f4775951c9c111e12cbee : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a67c21200f49fc9b44f111eb16e094 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05678c7cbab4458889856ab9c3829d8f : SUCCESS in 2m 36s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e598d916bbe44a36aa57a7254f20803a : SUCCESS in 31m 24s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/75706b052b9f4544b44d773439d90299 : SUCCESS in 48m 05s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/8b58427295934f389d4be4181e6f16b8 : SUCCESS in 3m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a73e2a1a57a6c5ede53fb2d3b4f6c6805202ba2c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-07-29 15:07:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87869b4dad5b026d750d047f9b97e1107188faf2","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-09-22 18:00:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7771f6317f69444f6572ab1105c54c6d8671871e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:01:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ee3f9a47026e144124ebfb0d14ab252fc4c8ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:52:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79f8335649946b08a6954d9ff294118\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b5f69317e6a4361a6dee0cbcce07af9 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7157ff5f51bf4a4587f7a20e948d9765 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/879672988ac9455591692fca135cd7ad : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/070a51ae53454198b6e07e480583b356 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f6f1ca3edf41df952ec5fb72f1c799 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0987f87d38c4696b6d47a9dff545fe8 : SUCCESS in 3m 31s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/08b80dd6b23a40b8b259176177330683 : SUCCESS in 44m 55s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/aa5cff381a58478092e6303887a7f1f8 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fcce77739e62916084fc517d752d4afc47178f44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:52:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a0e96b8b1714be1190257128605a95b8fedfe4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 18:55:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8772918725479498447a324d7efb34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f79558163a24616bf1ed847e43ef3df : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0cc0ccd464e3903b502e7d49b2cf90fd503fa3d","revisions":{"b0cc0ccd464e3903b502e7d49b2cf90fd503fa3d":{"kind":"REWORK","_number":1,"created":"2026-06-05 17:00:36.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/06/991906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/06/991906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/06/991906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/06/991906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/06/991906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/06/991906/1"}}},"commit":{"parents":[{"commit":"a55fd43427e594091f423aff6e2253a3678d50a1","subject":"Handle BIOS Firmware Update the same way as real hardware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a55fd43427e594091f423aff6e2253a3678d50a1"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2026-06-03 00:09:03.000000000","tz":0},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2026-06-05 17:00:21.000000000","tz":0},"subject":"Add ComputerSystem SerialNumber derived from domain UUID","message":"Add ComputerSystem SerialNumber derived from domain UUID\n\nChange-Id: I10885f643ae8642311600fe1c4839adb8f1f8c93\nSigned-off-by: s3rj1k \u003cevasive.gyron@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b0cc0ccd464e3903b502e7d49b2cf90fd503fa3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/b0cc0ccd464e3903b502e7d49b2cf90fd503fa3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
