)]}'
{"id":"openstack%2Fsushy-tools~991909","triplet_id":"openstack%2Fsushy-tools~master~I22520b31df098c5929746967630014a8d3996b64","project":"openstack/sushy-tools","branch":"master","attention_set":{"37442":{"account":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"last_update":"2026-07-30 10:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2026-07-30 10:59:32.000000000","reason":"\u003cGERRIT_ACCOUNT_10239\u003e replied on the change","reason_account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}},"hashtags":[],"change_id":"I22520b31df098c5929746967630014a8d3996b64","subject":"Add EthernetInterface UefiDevicePath derived from PCI topology","status":"NEW","created":"2026-06-05 17:00:43.000000000","updated":"2026-07-30 10:59:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"50f3a21f018b38fdf6400a4dc2c2f6e79bdb1b6b","_number":991909,"virtual_id_number":991909,"owner":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-29 17:03:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-30 10:59:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-05 17:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-29 15:16:58.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2026-07-30 10:59:32.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"6142b266a06bb572f498e631e9e7d063cfbaf8c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2026-06-05 17:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3edaa3b21a6a441a7c15e225277c4f50d24b8aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 17:56:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7caec600d97c4b43952f94f4b1460299\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e7e8dd1de54b069d2beffad338bc46 : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21efa89602724e1d89c224895d39b259 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c6c4a5a7260407ebde0ce0e13b2adfa : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb8bc38df28444d59a93edd238a41710 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e2cbfa793641f1a78cee6e07c749b8 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/341aed464ee54357b9e05aeb9ce3893e : SUCCESS in 3m 19s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/46220cd1a6a7435d832df3bbe8bf7faa : SUCCESS in 43m 12s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/01a2173ff62249b885541e748bdfa267 : SUCCESS in 49m 06s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/ddc76df9377e4c4c81359d45e592abce : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b586473648eb3dbb428bbef8a11b498a36e06d14","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-07-29 15:16:58.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a97428c21bdc1ddd0f0f27eb861cdad7a66a9294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2026-07-29 16:06:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"df4ae0ce5c65f988a8c4e8c2674ad5a0c93f7cf7","author":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"date":"2026-07-29 16:10:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7182e4f5b3cdb742ec9504033e03a1e08bb9031e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:03:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17cae42c919540829b238cc07f51fe7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e197f3c6eb9f4f9682dab4d3875dd586 : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37ce9d9468df4c2da08cd567f3774dfc : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faf58d80e06f48cab130e40ebb666453 : SUCCESS in 3m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e61476047b8644bd85cee5afd9f6edb0 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e79a972cc44e0cb42cd451c2d36532 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5836cf29a6de4891aa8eea4910c5d75f : SUCCESS in 3m 27s\n- sushy-tools-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/af8216887d704af99d0e58810327bfbd : SUCCESS in 42m 36s\n- sushy-tools-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/97159bd699374a82bda7d92f385f50a6 : SUCCESS in 48m 39s\n- sushy-tools-tox-codespell https://zuul.opendev.org/t/openstack/build/b681b9468dd84c9aa9a0790ee0f35c55 : SUCCESS in 2m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50f3a21f018b38fdf6400a4dc2c2f6e79bdb1b6b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2026-07-30 10:59:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"88a9c6a07dbbf4fc138183fb0da0c1a45783ec72","revisions":{"52417b76a272f9e21fffca2618d5a653472674e9":{"kind":"REWORK","_number":1,"created":"2026-06-05 17:00:43.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/09/991909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/09/991909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/1"}}},"commit":{"parents":[{"commit":"a55fd43427e594091f423aff6e2253a3678d50a1","subject":"Handle BIOS Firmware Update the same way as real hardware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a55fd43427e594091f423aff6e2253a3678d50a1"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2026-06-03 14:26:28.000000000","tz":0},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2026-06-05 17:00:21.000000000","tz":0},"subject":"Add EthernetInterface UefiDevicePath derived from PCI topology","message":"Add EthernetInterface UefiDevicePath derived from PCI topology\n\nChange-Id: I22520b31df098c5929746967630014a8d3996b64\nSigned-off-by: s3rj1k \u003cevasive.gyron@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/52417b76a272f9e21fffca2618d5a653472674e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/52417b76a272f9e21fffca2618d5a653472674e9"}]},"branch":"refs/heads/master"},"88a9c6a07dbbf4fc138183fb0da0c1a45783ec72":{"kind":"REWORK","_number":2,"created":"2026-07-29 16:06:40.000000000","uploader":{"_account_id":37442,"name":"Serhii Ivanov","email":"evasive.gyron@gmail.com","username":"s3rj1k"},"ref":"refs/changes/09/991909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy-tools","ref":"refs/changes/09/991909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy-tools refs/changes/09/991909/2"}}},"commit":{"parents":[{"commit":"a55fd43427e594091f423aff6e2253a3678d50a1","subject":"Handle BIOS Firmware Update the same way as real hardware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/a55fd43427e594091f423aff6e2253a3678d50a1"}]}],"author":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2026-06-03 14:26:28.000000000","tz":0},"committer":{"name":"s3rj1k","email":"evasive.gyron@gmail.com","date":"2026-07-29 16:06:19.000000000","tz":0},"subject":"Add EthernetInterface UefiDevicePath derived from PCI topology","message":"Add EthernetInterface UefiDevicePath derived from PCI topology\n\nChange-Id: I22520b31df098c5929746967630014a8d3996b64\nSigned-off-by: s3rj1k \u003cevasive.gyron@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/88a9c6a07dbbf4fc138183fb0da0c1a45783ec72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy-tools/commit/88a9c6a07dbbf4fc138183fb0da0c1a45783ec72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
