)]}'
{"id":"openstack%2Fsushy~651560","triplet_id":"openstack%2Fsushy~master~Iaabdda6c1fe0c0c492ad05c19f34172864703da8","project":"openstack/sushy","branch":"master","topic":"add-power","hashtags":[],"change_id":"Iaabdda6c1fe0c0c492ad05c19f34172864703da8","subject":"Adding Power resource schema","status":"MERGED","created":"2019-04-10 14:37:24.000000000","updated":"2019-04-26 14:47:44.000000000","submitted":"2019-04-26 14:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651560-1556290064216-99a4d632","meta_rev_id":"c203fe80771a3b682ef9fead9728b916a60a900b","_number":651560,"virtual_id_number":651560,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"date":"2019-04-26 14:22:10.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-04-26 14:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2019-04-17 14:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-26 13:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-26 13:52:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 14:16:04.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-04-26 13:52:25.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-04-26 14:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07ef82c5b41a4bba5b352c496749e7fbdec391d2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-10 14:37:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcab3a68d6e50a9dcaa7d0fab4387fa6d7af2f3f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-10 14:38:00.000000000","message":"Patch Set 1: Workflow-1\n\nmissing unit test, release note and moarrrr stuff","accounts_in_message":[],"_revision_number":1},{"id":"d69f8e9ccf7bcfacf3e1100df83a8d884afd8dd1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-10 15:14:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"be4d3596a555523ac1deba640d680aa6f0242437","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 16:09:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/60/651560/2/check/openstack-tox-cover/3ca024b/cover/ : SUCCESS in 4m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/651560/2/check/openstack-tox-lower-constraints/22142af/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/60/651560/2/check/openstack-tox-pep8/d675f3e/ : FAILURE in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/60/651560/2/check/openstack-tox-py27/c1e68ee/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/60/651560/2/check/openstack-tox-py35/0a8860f/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/60/651560/2/check/openstack-tox-py36/3f9a395/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/60/651560/2/check/openstack-tox-docs/acdd7e8/html/ : SUCCESS in 6m 25s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/651560/2/check/sushy-tempest-ironic-partition-redfish-src/01969fe/ : SUCCESS in 46m 15s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/651560/2/check/sushy-tempest-ironic-partition-redfish-src-python2/61e554a/ : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a2dc6f057e720bfb2d1b803f8664e0139f7d4f82","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-11 10:04:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3951c4c3186f7ed5e2c4eb7008af2e7e42bd461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 10:54:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/60/651560/3/check/openstack-tox-cover/b7c7c3a/cover/ : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/651560/3/check/openstack-tox-lower-constraints/b253a66/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/60/651560/3/check/openstack-tox-pep8/4318765/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/60/651560/3/check/openstack-tox-py27/bd58db8/ : SUCCESS in 5m 03s\n- openstack-tox-py35 http://logs.openstack.org/60/651560/3/check/openstack-tox-py35/9e3c573/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/60/651560/3/check/openstack-tox-py36/7df203c/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/60/651560/3/check/openstack-tox-docs/db9b7e5/html/ : SUCCESS in 5m 00s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/651560/3/check/sushy-tempest-ironic-partition-redfish-src/f550377/ : SUCCESS in 41m 39s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/651560/3/check/sushy-tempest-ironic-partition-redfish-src-python2/efd622c/ : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ad8e2ca88f2f3590445d6cda64a9954a0a490a75","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-11 11:58:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa7a6adaf88ed71265214d09df66f4d7309e4559","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 12:44:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/60/651560/4/check/openstack-tox-cover/1c552fa/cover/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/651560/4/check/openstack-tox-lower-constraints/74fec9a/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/60/651560/4/check/openstack-tox-pep8/9c96d4f/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/60/651560/4/check/openstack-tox-py27/a1e8ff7/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/60/651560/4/check/openstack-tox-py35/5143aea/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/60/651560/4/check/openstack-tox-py36/05af5ac/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/60/651560/4/check/openstack-tox-docs/0e80d99/html/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/60/651560/4/check/build-openstack-releasenotes/a188d96/html/ : SUCCESS in 3m 59s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/651560/4/check/sushy-tempest-ironic-partition-redfish-src/a795a84/ : SUCCESS in 39m 17s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/651560/4/check/sushy-tempest-ironic-partition-redfish-src-python2/59c8834/ : SUCCESS in 40m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a9e3d83eba69c4b2fb57ed1ad63a409eaecbfb9c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-11 14:23:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0cca258a81b1ae935229f57206fda714b712faf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 15:28:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/60/651560/5/check/openstack-tox-cover/05447ac/cover/ : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/651560/5/check/openstack-tox-lower-constraints/73cfd0c/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/60/651560/5/check/openstack-tox-pep8/8b6e979/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/60/651560/5/check/openstack-tox-py27/2a0bfbf/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/60/651560/5/check/openstack-tox-py35/db0286b/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/60/651560/5/check/openstack-tox-py36/bd06b15/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/60/651560/5/check/openstack-tox-docs/51d6237/html/ : SUCCESS in 4m 08s\n- build-openstack-releasenotes http://logs.openstack.org/60/651560/5/check/build-openstack-releasenotes/ebf2174/html/ : SUCCESS in 3m 37s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/651560/5/check/sushy-tempest-ironic-partition-redfish-src/c5f70a4/ : SUCCESS in 57m 50s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/651560/5/check/sushy-tempest-ironic-partition-redfish-src-python2/03df439/ : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":5},{"id":"a47c4fb1f8212e86a9a88d4fdabb392dcf390a6b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-04-17 14:16:04.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b7ef93645c98665ad6d48c183d8eb3bf083dd9e5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-04-26 13:52:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nI guess this new resource needs wiring into Chassis? Can be done in a follow-up.","accounts_in_message":[],"_revision_number":5},{"id":"8602af84d976fa8e677f657ef761f10b1b07c63c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 13:52:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2fcf9a6836e7d06f97c0b641463681a989fafa99","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-26 14:22:10.000000000","message":"Patch Set 5:\n\nyes, both this and Thermal will need to be linked to chassis","accounts_in_message":[],"_revision_number":5},{"id":"fb2627e64404627e45a21ea09952b89466ef32df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 14:47:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/60/651560/5/gate/openstack-tox-lower-constraints/7573bba/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/60/651560/5/gate/openstack-tox-pep8/f6313d0/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/60/651560/5/gate/openstack-tox-py27/c28223c/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/60/651560/5/gate/openstack-tox-py36/1d6e429/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/60/651560/5/gate/openstack-tox-docs/fe691d7/html/ : SUCCESS in 4m 15s\n- build-openstack-releasenotes http://logs.openstack.org/60/651560/5/gate/build-openstack-releasenotes/af409b2/html/ : SUCCESS in 4m 03s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/651560/5/gate/sushy-tempest-ironic-partition-redfish-src/99274cd/ : SUCCESS in 48m 56s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/651560/5/gate/sushy-tempest-ironic-partition-redfish-src-python2/b6e72e2/ : SUCCESS in 46m 33s","accounts_in_message":[],"_revision_number":5},{"id":"415c7f939f22688d7252c1dd1574d46d12705e8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 14:47:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ad15a88908c64bd4acafdec477ba3c1a5aff0420","revisions":{"187b0cc5c3e79bc084b09f171ffc5e7a435c3057":{"kind":"REWORK","_number":1,"created":"2019-04-10 14:37:24.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/651560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/651560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/651560/1"}}},"commit":{"parents":[{"commit":"9a138d4ec9f61e3145fc29606082f086c367a53d","subject":"Merge \"Add versions to release notes series\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a138d4ec9f61e3145fc29606082f086c367a53d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 14:36:22.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 14:36:22.000000000","tz":120},"subject":"[WIP] Adding Power resource schema","message":"[WIP] Adding Power resource schema\n\nAdding Power resource schema from Redfish API\n\nChange-Id: Iaabdda6c1fe0c0c492ad05c19f34172864703da8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/187b0cc5c3e79bc084b09f171ffc5e7a435c3057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/187b0cc5c3e79bc084b09f171ffc5e7a435c3057"}]},"branch":"refs/heads/master"},"fe8264c39eb86f03f6c8dbc0bb2ac550dd77d183":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-10 15:14:53.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/651560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/651560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/651560/2"}}},"commit":{"parents":[{"commit":"9a138d4ec9f61e3145fc29606082f086c367a53d","subject":"Merge \"Add versions to release notes series\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a138d4ec9f61e3145fc29606082f086c367a53d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 14:36:22.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 15:14:10.000000000","tz":120},"subject":"[WIP] Adding Power resource schema","message":"[WIP] Adding Power resource schema\n\nAdding Power resource schema from Redfish API\n\nChange-Id: Iaabdda6c1fe0c0c492ad05c19f34172864703da8\nStory: 2005414\nTask: 30451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fe8264c39eb86f03f6c8dbc0bb2ac550dd77d183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fe8264c39eb86f03f6c8dbc0bb2ac550dd77d183"}]},"branch":"refs/heads/master"},"b486c4e22e39022a6a60d30f94680b525a2f8e50":{"kind":"REWORK","_number":3,"created":"2019-04-11 10:04:23.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/651560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/651560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/651560/3"}}},"commit":{"parents":[{"commit":"9a138d4ec9f61e3145fc29606082f086c367a53d","subject":"Merge \"Add versions to release notes series\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a138d4ec9f61e3145fc29606082f086c367a53d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 14:36:22.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-11 10:04:12.000000000","tz":120},"subject":"[WIP] Adding Power resource schema","message":"[WIP] Adding Power resource schema\n\nAdding Power resource schema from Redfish API\n\nChange-Id: Iaabdda6c1fe0c0c492ad05c19f34172864703da8\nStory: 2005414\nTask: 30451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b486c4e22e39022a6a60d30f94680b525a2f8e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b486c4e22e39022a6a60d30f94680b525a2f8e50"}]},"branch":"refs/heads/master"},"5d228035bd527221be08977d0a0c68273acf9012":{"kind":"REWORK","_number":4,"created":"2019-04-11 11:58:30.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/651560/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/651560/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/651560/4"}}},"commit":{"parents":[{"commit":"9a138d4ec9f61e3145fc29606082f086c367a53d","subject":"Merge \"Add versions to release notes series\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a138d4ec9f61e3145fc29606082f086c367a53d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 14:36:22.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-11 11:58:21.000000000","tz":120},"subject":"[WIP] Adding Power resource schema","message":"[WIP] Adding Power resource schema\n\nAdding Power resource schema from Redfish API\n\nChange-Id: Iaabdda6c1fe0c0c492ad05c19f34172864703da8\nStory: 2005414\nTask: 30451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d228035bd527221be08977d0a0c68273acf9012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d228035bd527221be08977d0a0c68273acf9012"}]},"branch":"refs/heads/master"},"ad15a88908c64bd4acafdec477ba3c1a5aff0420":{"kind":"REWORK","_number":5,"created":"2019-04-11 14:23:57.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/60/651560/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/651560/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/651560/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/651560/5"}}},"commit":{"parents":[{"commit":"9a138d4ec9f61e3145fc29606082f086c367a53d","subject":"Merge \"Add versions to release notes series\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a138d4ec9f61e3145fc29606082f086c367a53d"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-10 14:36:22.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-11 14:23:41.000000000","tz":120},"subject":"Adding Power resource schema","message":"Adding Power resource schema\n\nAdding Power resource schema from Redfish API\n\nChange-Id: Iaabdda6c1fe0c0c492ad05c19f34172864703da8\nStory: 2005414\nTask: 30451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ad15a88908c64bd4acafdec477ba3c1a5aff0420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ad15a88908c64bd4acafdec477ba3c1a5aff0420"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
