)]}'
{"id":"openstack%2Fsushy~655247","triplet_id":"openstack%2Fsushy~master~I76c99da2aac8e0629bb468e22abd104cb29bc4ab","project":"openstack/sushy","branch":"master","topic":"fabric-endpoint-subresource","hashtags":[],"change_id":"I76c99da2aac8e0629bb468e22abd104cb29bc4ab","subject":"Add ``Endpoint`` sub-resource to ``Fabric``","status":"MERGED","created":"2019-04-23 20:44:22.000000000","updated":"2019-07-01 12:44:38.000000000","submitted":"2019-07-01 12:44:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655247-1561985078165-971b6051","meta_rev_id":"a1eba6b756302c488e20abdc08704b7a2d97eeca","_number":655247,"virtual_id_number":655247,"owner":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-07-01 12:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-21 12:48:34.000000000","_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2019-06-26 06:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-06-21 13:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2019-06-26 06:58:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 13:58:33.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-06-26 06:58:36.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-07-01 12:44:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b62c4ee4be6dc1664ceab7d4e51ceb13ee1b4cf","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-04-23 20:44:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61e72ea0904cbfec082d037c0d8d4b8fee24933f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 21:29:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/47/655247/1/check/openstack-tox-cover/7e12aed/cover/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655247/1/check/openstack-tox-lower-constraints/f3b07c7/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/47/655247/1/check/openstack-tox-pep8/5a4996e/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/47/655247/1/check/openstack-tox-py27/d556c58/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/47/655247/1/check/openstack-tox-py36/7e5ac8b/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/47/655247/1/check/openstack-tox-docs/8279555/html/ : SUCCESS in 5m 57s\n- build-openstack-releasenotes http://logs.openstack.org/47/655247/1/check/build-openstack-releasenotes/881985d/html/ : SUCCESS in 3m 47s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/47/655247/1/check/sushy-tempest-ironic-partition-redfish-src/8b9acc7/ : FAILURE in 8m 33s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/47/655247/1/check/sushy-tempest-ironic-partition-redfish-src-python2/90fad9a/ : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"bc0886b2cf941b0164821d2a2463a3c7337aaa9d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 08:20:32.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nthanks for the patch!","accounts_in_message":[],"_revision_number":1},{"id":"e79eadeb7ba20e86daef9524cea0b8922a9efc44","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-04-24 09:37:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"882eb3eda42c4175c8447c6bfc271f261459850f","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-04-24 09:40:37.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e thanks for the patch!","accounts_in_message":[],"_revision_number":2},{"id":"977fb88b4ac7f17e296a3cc2f0624bf06e58c4fb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 10:04:40.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nlooks good, minor comments inline","accounts_in_message":[],"_revision_number":2},{"id":"6ca2a4bd37b6c2860e0c139bb6a11c26445d0dcc","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-04-24 10:12:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"794513f32d6c5c2a18b3917ff1e8719ff4520410","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-04-24 10:15:50.000000000","message":"Patch Set 3:\n\n(3 comments)\n\n\u003e (4 comments)\n \u003e \n \u003e looks good, minor comments inline","accounts_in_message":[],"_revision_number":3},{"id":"51cfd90917d3451221e8c0ae83c633b0ac26bf37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:06:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/47/655247/3/check/openstack-tox-cover/c5f9b9f/cover/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655247/3/check/openstack-tox-lower-constraints/ceb51aa/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/47/655247/3/check/openstack-tox-pep8/d9bb050/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/47/655247/3/check/openstack-tox-py27/fb9cf77/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/47/655247/3/check/openstack-tox-py36/da69c2c/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/47/655247/3/check/openstack-tox-docs/891280d/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/47/655247/3/check/build-openstack-releasenotes/499887b/html/ : SUCCESS in 4m 08s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/47/655247/3/check/sushy-tempest-ironic-partition-redfish-src/907acc0/ : SUCCESS in 46m 04s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/47/655247/3/check/sushy-tempest-ironic-partition-redfish-src-python2/eda7630/ : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":3},{"id":"099cbab4cf2ea518af0d986ff97cfea2abd4a6e9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 18:41:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1db4cc9ea90813559f6f00a22fe99f234e6ce4b6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-06-20 22:11:50.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"570870faa194e194da0be4a9fcc7757ed64e9eda","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-06-21 12:46:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5443cb96a72af6bd2a8f3d2271c40adadec9205","author":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"date":"2019-06-21 12:48:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60df25bc611ddd0ca576493af50b8c185347c0db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 13:42:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/47/655247/4/check/openstack-tox-cover/c25718e/cover/ : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655247/4/check/openstack-tox-lower-constraints/65a2ab8/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/47/655247/4/check/openstack-tox-pep8/b701ce2/ : SUCCESS in 9m 47s\n- openstack-tox-py27 http://logs.openstack.org/47/655247/4/check/openstack-tox-py27/113e074/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/47/655247/4/check/openstack-tox-py36/72aa947/ : SUCCESS in 7m 39s\n- openstack-tox-py37 http://logs.openstack.org/47/655247/4/check/openstack-tox-py37/3a1c14f/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/47/655247/4/check/openstack-tox-docs/79022a7/html/ : SUCCESS in 5m 21s\n- build-openstack-releasenotes http://logs.openstack.org/47/655247/4/check/build-openstack-releasenotes/63440d7/html/ : SUCCESS in 4m 36s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/47/655247/4/check/sushy-tempest-ironic-partition-redfish-src/c46a8c7/ : SUCCESS in 48m 54s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/47/655247/4/check/sushy-tempest-ironic-partition-redfish-src-python2/3c4879b/ : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":4},{"id":"fd54e8d4f71b53f027dd7d7639da94a97e3b5812","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-21 13:58:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"17052f84b1c504abca8f8c38b64b103ddd540943","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-06-26 06:58:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"40eaa65cca0d59c58aba6dc2d68aaf41470c54d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 06:58:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eac70da9bf27d8b1fd68ea04b7a485894883e6bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 07:38:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655247/4/gate/openstack-tox-lower-constraints/4fd305a/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/47/655247/4/gate/openstack-tox-pep8/fd383fc/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/47/655247/4/gate/openstack-tox-py27/1ad02d3/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/47/655247/4/gate/openstack-tox-py36/35f4bd7/ : SUCCESS in 3m 06s\n- openstack-tox-py37 http://logs.openstack.org/47/655247/4/gate/openstack-tox-py37/8393420/ : SUCCESS in 3m 00s\n- openstack-tox-docs http://logs.openstack.org/47/655247/4/gate/openstack-tox-docs/5810469/html/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/47/655247/4/gate/build-openstack-releasenotes/4842c1c/html/ : SUCCESS in 3m 13s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/47/655247/4/gate/sushy-tempest-ironic-partition-redfish-src/83e3711/ : FAILURE in 33m 13s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/47/655247/4/gate/sushy-tempest-ironic-partition-redfish-src-python2/d03a1dd/ : FAILURE in 33m 17s","accounts_in_message":[],"_revision_number":4},{"id":"aecd36ee0e3a9eb841dae795b7fb344a1b036707","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-01 11:02:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"187b0692438029df50b7b4332d3abd71f43919f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:49:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/47/655247/4/check/openstack-tox-cover/420a381/cover/ : SUCCESS in 2m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655247/4/check/openstack-tox-lower-constraints/cce94c9/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/47/655247/4/check/openstack-tox-pep8/8793616/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/47/655247/4/check/openstack-tox-py27/0b029a7/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/47/655247/4/check/openstack-tox-py36/59e457d/ : SUCCESS in 2m 43s\n- openstack-tox-py37 http://logs.openstack.org/47/655247/4/check/openstack-tox-py37/1523536/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/47/655247/4/check/openstack-tox-docs/a747136/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/47/655247/4/check/build-openstack-releasenotes/0dd697e/html/ : SUCCESS in 3m 08s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/47/655247/4/check/sushy-tempest-ironic-partition-redfish-src/e30d39b/ : SUCCESS in 42m 51s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/47/655247/4/check/sushy-tempest-ironic-partition-redfish-src-python2/10b5e7b/ : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":4},{"id":"a80400c881ed233d0d7283657ba18bc6742e0cfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:49:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"91a6e9bcde09ec881a2dd662d37d8621199ba4ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 12:44:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/47/655247/4/gate/openstack-tox-lower-constraints/4c8e46e/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/47/655247/4/gate/openstack-tox-pep8/14b413d/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/47/655247/4/gate/openstack-tox-py27/2f57d35/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/47/655247/4/gate/openstack-tox-py36/dec783a/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/47/655247/4/gate/openstack-tox-py37/039853a/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/47/655247/4/gate/openstack-tox-docs/8cb33d7/html/ : SUCCESS in 2m 47s\n- build-openstack-releasenotes http://logs.openstack.org/47/655247/4/gate/build-openstack-releasenotes/4aa0150/html/ : SUCCESS in 4m 03s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/47/655247/4/gate/sushy-tempest-ironic-partition-redfish-src/9046f03/ : SUCCESS in 52m 52s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/47/655247/4/gate/sushy-tempest-ironic-partition-redfish-src-python2/a69b74e/ : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5358f94a2fb0450178a8bd22a049ee1f2d975dbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 12:44:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ca8471196ea86e8f32e0676c5bfaf979c749479c","revisions":{"99b88176ddd3ae9cc899b52da6c07eb7f6b66eaa":{"kind":"REWORK","_number":1,"created":"2019-04-23 20:44:22.000000000","uploader":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"ref":"refs/changes/47/655247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/47/655247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/47/655247/1"}}},"commit":{"parents":[{"commit":"f10322a67c558d938a655e621934489bb9bcf09e","subject":"Merge \"Add `FabricCollection` and `Fabric` classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f10322a67c558d938a655e621934489bb9bcf09e"}]}],"author":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-23 20:36:31.000000000","tz":330},"committer":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-23 20:42:04.000000000","tz":330},"subject":"Add ``Endpoint`` sub-resource to ``Fabric``","message":"Add ``Endpoint`` sub-resource to ``Fabric``\n\nAdd representation of the Endpoint sub-resource to Fabric resource in sushy.\nA fabric endpoint represents the properties of an entity that sends or receives\nprotocol defined messages over a transport.\n\nStory: #2003853\nTask: #26648\n\nChange-Id: I76c99da2aac8e0629bb468e22abd104cb29bc4ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/99b88176ddd3ae9cc899b52da6c07eb7f6b66eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/99b88176ddd3ae9cc899b52da6c07eb7f6b66eaa"}]},"branch":"refs/heads/master"},"e20e627823b7ae416e17ed10723db1abb87d58fb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-24 09:37:54.000000000","uploader":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"ref":"refs/changes/47/655247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/47/655247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/47/655247/2"}}},"commit":{"parents":[{"commit":"f10322a67c558d938a655e621934489bb9bcf09e","subject":"Merge \"Add `FabricCollection` and `Fabric` classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f10322a67c558d938a655e621934489bb9bcf09e"}]}],"author":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-23 20:36:31.000000000","tz":330},"committer":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-24 09:31:49.000000000","tz":330},"subject":"Add ``Endpoint`` sub-resource to ``Fabric``","message":"Add ``Endpoint`` sub-resource to ``Fabric``\n\nAdd representation of the Endpoint sub-resource to Fabric resource\nin sushy. A fabric endpoint represents the properties of an entity\nthat sends or receives protocol defined messages over a transport.\n\nStory: #2003853\nTask: #26648\n\nChange-Id: I76c99da2aac8e0629bb468e22abd104cb29bc4ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e20e627823b7ae416e17ed10723db1abb87d58fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e20e627823b7ae416e17ed10723db1abb87d58fb"}]},"branch":"refs/heads/master"},"c6da336f7598234215b091aebb9e917dd4706687":{"kind":"REWORK","_number":3,"created":"2019-04-24 10:12:49.000000000","uploader":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"ref":"refs/changes/47/655247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/47/655247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/47/655247/3"}}},"commit":{"parents":[{"commit":"f10322a67c558d938a655e621934489bb9bcf09e","subject":"Merge \"Add `FabricCollection` and `Fabric` classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f10322a67c558d938a655e621934489bb9bcf09e"}]}],"author":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-23 20:36:31.000000000","tz":330},"committer":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-24 10:12:01.000000000","tz":330},"subject":"Add ``Endpoint`` sub-resource to ``Fabric``","message":"Add ``Endpoint`` sub-resource to ``Fabric``\n\nAdd representation of the Endpoint sub-resource to Fabric resource\nin sushy. A fabric endpoint represents the properties of an entity\nthat sends or receives protocol defined messages over a transport.\n\nStory: #2003853\nTask: #26648\n\nChange-Id: I76c99da2aac8e0629bb468e22abd104cb29bc4ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c6da336f7598234215b091aebb9e917dd4706687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c6da336f7598234215b091aebb9e917dd4706687"}]},"branch":"refs/heads/master"},"ca8471196ea86e8f32e0676c5bfaf979c749479c":{"kind":"REWORK","_number":4,"created":"2019-06-21 12:46:16.000000000","uploader":{"_account_id":30013,"name":"Varsha Verma","email":"varsha.verma.eee15@itbhu.ac.in","username":"verma-varsha"},"ref":"refs/changes/47/655247/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/47/655247/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/655247/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/47/655247/4"}}},"commit":{"parents":[{"commit":"f10322a67c558d938a655e621934489bb9bcf09e","subject":"Merge \"Add `FabricCollection` and `Fabric` classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f10322a67c558d938a655e621934489bb9bcf09e"}]}],"author":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-04-23 20:36:31.000000000","tz":330},"committer":{"name":"Varsha","email":"varsha.verma.eee15@itbhu.ac.in","date":"2019-06-21 12:45:41.000000000","tz":330},"subject":"Add ``Endpoint`` sub-resource to ``Fabric``","message":"Add ``Endpoint`` sub-resource to ``Fabric``\n\nAdd representation of the Endpoint sub-resource to Fabric resource\nin sushy. A fabric endpoint represents the properties of an entity\nthat sends or receives protocol defined messages over a transport.\n\nStory: #2003853\nTask: #26648\n\nChange-Id: I76c99da2aac8e0629bb468e22abd104cb29bc4ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ca8471196ea86e8f32e0676c5bfaf979c749479c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ca8471196ea86e8f32e0676c5bfaf979c749479c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
