)]}'
{"id":"openstack%2Fsushy~656229","triplet_id":"openstack%2Fsushy~master~I0f1e1cb920268f059062c12a6a08d25cb905f680","project":"openstack/sushy","branch":"master","hashtags":[],"change_id":"I0f1e1cb920268f059062c12a6a08d25cb905f680","subject":"update git.openstack.org to opendev","status":"MERGED","created":"2019-04-29 02:41:53.000000000","updated":"2019-04-29 09:41:41.000000000","submitted":"2019-04-29 09:41:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656229-1556530901129-0411a076","meta_rev_id":"930087fa2ad31190fe8942f39738e6278c90a0fa","_number":656229,"virtual_id_number":656229,"owner":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2019-04-29 09:41:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-04-29 08:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-04-29 08:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2019-04-29 08:56:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 08:53:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-04-29 08:56:25.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-04-29 09:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eac8bac92a799f287beaabcbe7588e8202aa029b","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-29 02:41:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98f3392cba10b77c703d9da9ed9c0fef71024c77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 03:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/656229/1/check/test-release-openstack/937b025/ : SUCCESS in 3m 12s\n- openstack-tox-cover http://logs.openstack.org/29/656229/1/check/openstack-tox-cover/ad18698/cover/ : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/656229/1/check/openstack-tox-lower-constraints/2079cc3/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/29/656229/1/check/openstack-tox-pep8/91fba9c/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/29/656229/1/check/openstack-tox-py27/8634345/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/29/656229/1/check/openstack-tox-py36/ee1f8a4/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/29/656229/1/check/openstack-tox-docs/930d236/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/29/656229/1/check/build-openstack-releasenotes/4962c8b/html/ : SUCCESS in 3m 39s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/29/656229/1/check/sushy-tempest-ironic-partition-redfish-src/49f0e3e/ : SUCCESS in 45m 00s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/29/656229/1/check/sushy-tempest-ironic-partition-redfish-src-python2/2467131/ : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":1},{"id":"89b6f7020671395d3ba75d09842c6ed38d0a0f3a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-29 08:53:57.000000000","message":"Patch Set 1: Code-Review+2\n\nlooks good, thanks","accounts_in_message":[],"_revision_number":1},{"id":"df4d9c0d01c8bae172de96abe281efbad7d37c33","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-04-29 08:56:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5389fef80bf59ea5f28f4a1ac2127f2068ee09f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 08:56:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11d91ad34fc83f2cc77e2de279c80204fe0ba737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 09:41:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/29/656229/1/gate/test-release-openstack/0ac775f/ : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/29/656229/1/gate/openstack-tox-lower-constraints/7b08b95/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/29/656229/1/gate/openstack-tox-pep8/1ad5e2c/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/29/656229/1/gate/openstack-tox-py27/18dcf8b/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/29/656229/1/gate/openstack-tox-py36/b37da5d/ : SUCCESS in 5m 08s\n- openstack-tox-docs http://logs.openstack.org/29/656229/1/gate/openstack-tox-docs/161e197/html/ : SUCCESS in 3m 50s\n- build-openstack-releasenotes http://logs.openstack.org/29/656229/1/gate/build-openstack-releasenotes/564a3c1/html/ : SUCCESS in 4m 17s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/29/656229/1/gate/sushy-tempest-ironic-partition-redfish-src/bb5bd5d/ : SUCCESS in 40m 15s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/29/656229/1/gate/sushy-tempest-ironic-partition-redfish-src-python2/7c9b3f3/ : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"930087fa2ad31190fe8942f39738e6278c90a0fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 09:41:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"103737ad3cab03a0f6c148b8aef47699928febe4","revisions":{"103737ad3cab03a0f6c148b8aef47699928febe4":{"kind":"REWORK","_number":1,"created":"2019-04-29 02:41:53.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/29/656229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/29/656229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/29/656229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/29/656229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/29/656229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/29/656229/1"}}},"commit":{"parents":[{"commit":"775f6b6f994e7bf5e0097f00fc82e435fffc8ebb","subject":"Merge \"Adding Thermal resource schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/775f6b6f994e7bf5e0097f00fc82e435fffc8ebb"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-29 02:41:28.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-29 02:41:50.000000000","tz":480},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: I0f1e1cb920268f059062c12a6a08d25cb905f680\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/103737ad3cab03a0f6c148b8aef47699928febe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/103737ad3cab03a0f6c148b8aef47699928febe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
