)]}'
{"id":"openstack%2Fsushy~656965","triplet_id":"openstack%2Fsushy~master~I041a896f01503ddfac7f343e36b10b9be5318e9f","project":"openstack/sushy","branch":"master","topic":"chassis-thermal-power","hashtags":[],"change_id":"I041a896f01503ddfac7f343e36b10b9be5318e9f","subject":"Add Power and Thermal resources to Chassis","status":"MERGED","created":"2019-05-03 15:35:02.000000000","updated":"2019-05-17 16:13:57.000000000","submitted":"2019-05-17 16:13:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656965-1558109637548-a5bd43bf","meta_rev_id":"e4a02f32b91046df74db3fbd1fa9742a3dd4041c","_number":656965,"virtual_id_number":656965,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2019-05-17 14:07:12.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-05-17 16:13:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2019-05-16 06:46:01.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-15 14:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2019-05-16 06:46:01.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 14:48:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-05-16 06:46:01.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-05-17 16:13:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"555bd0e0baa3dc79ca5ccdd789dc915b9d10a339","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-03 15:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7c5ed4c6821cae785ae7304f19c2dd4a8a15385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 16:25:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/65/656965/1/check/openstack-tox-cover/7b21070/cover/ : SUCCESS in 6m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/656965/1/check/openstack-tox-lower-constraints/138ac96/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/65/656965/1/check/openstack-tox-pep8/a276bd3/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/65/656965/1/check/openstack-tox-py27/97daa6e/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/65/656965/1/check/openstack-tox-py36/50845b3/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/65/656965/1/check/openstack-tox-docs/ca44daa/html/ : SUCCESS in 6m 53s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/65/656965/1/check/sushy-tempest-ironic-partition-redfish-src/fb3ddac/ : SUCCESS in 47m 20s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/65/656965/1/check/sushy-tempest-ironic-partition-redfish-src-python2/6972b75/ : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e9b76f4ac43be91809f4703d6f8ac560d0cab159","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-05-15 14:48:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"52fc267a5fdf84a44e9c2faf6d23035f37f5c4d8","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-05-16 06:46:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you.","accounts_in_message":[],"_revision_number":1},{"id":"0a7aa750582567113a8f40c74e912088da2be619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:46:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7705ba847b93a6e3372a52bcd7508b1e6fdf8254","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 07:37:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/656965/1/gate/openstack-tox-lower-constraints/0cf7787/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/65/656965/1/gate/openstack-tox-pep8/a9bb80d/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/65/656965/1/gate/openstack-tox-py27/1b394d4/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/65/656965/1/gate/openstack-tox-py36/181c2a2/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/65/656965/1/gate/openstack-tox-docs/b725b5b/html/ : SUCCESS in 4m 42s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/65/656965/1/gate/sushy-tempest-ironic-partition-redfish-src/89f2e9b/ : FAILURE in 45m 38s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/65/656965/1/gate/sushy-tempest-ironic-partition-redfish-src-python2/eb3ee99/ : FAILURE in 40m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3ed196ea70e3c1e1a7fc167160c3a25323eb7ab6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-16 07:46:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac404e2dd9e25850aa17d7cbc2f56fd0102670dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 08:20:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/65/656965/1/check/openstack-tox-cover/7e235dc/cover/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/656965/1/check/openstack-tox-lower-constraints/3041654/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/65/656965/1/check/openstack-tox-pep8/ab6a753/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/65/656965/1/check/openstack-tox-py27/9bd5769/ : SUCCESS in 6m 59s\n- openstack-tox-py36 http://logs.openstack.org/65/656965/1/check/openstack-tox-py36/ce701a8/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/65/656965/1/check/openstack-tox-docs/0d20371/html/ : SUCCESS in 6m 55s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/65/656965/1/check/sushy-tempest-ironic-partition-redfish-src/3374955/ : FAILURE in 32m 45s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/65/656965/1/check/sushy-tempest-ironic-partition-redfish-src-python2/f4761f8/ : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"23a4939e1eba695c926a7e8870abeaf778b5cbcf","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-05-17 14:07:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a824ff2cf342c6df614ad0a2e8708eee13966535","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:54:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/65/656965/1/check/openstack-tox-cover/cf08912/cover/ : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/656965/1/check/openstack-tox-lower-constraints/0be0f0b/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/65/656965/1/check/openstack-tox-pep8/8073218/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/65/656965/1/check/openstack-tox-py27/f6ce102/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/65/656965/1/check/openstack-tox-py36/2cbb824/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/65/656965/1/check/openstack-tox-docs/1cdd21a/html/ : SUCCESS in 7m 02s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/65/656965/1/check/sushy-tempest-ironic-partition-redfish-src/5c3d3ca/ : SUCCESS in 41m 35s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/65/656965/1/check/sushy-tempest-ironic-partition-redfish-src-python2/e45faf9/ : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cb81ed39f35e9781c7ad439fa4e2718e393b3b1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 14:54:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c13b774a938b6eaed05ff9b739fdaed5f85f0f14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 16:13:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/65/656965/1/gate/openstack-tox-lower-constraints/133cb0a/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/65/656965/1/gate/openstack-tox-pep8/d6d7bfe/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/65/656965/1/gate/openstack-tox-py27/965b0a7/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/65/656965/1/gate/openstack-tox-py36/bd4d1b3/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/65/656965/1/gate/openstack-tox-docs/1183efb/html/ : SUCCESS in 4m 21s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/65/656965/1/gate/sushy-tempest-ironic-partition-redfish-src/3b247cc/ : SUCCESS in 43m 50s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/65/656965/1/gate/sushy-tempest-ironic-partition-redfish-src-python2/1892109/ : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fc48fb0b312e7161983c24f32cb0b44faa6ca077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 16:13:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"188a72cee434f4f2cdabd5d71f82e97b28112b2e","revisions":{"188a72cee434f4f2cdabd5d71f82e97b28112b2e":{"kind":"REWORK","_number":1,"created":"2019-05-03 15:35:02.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/65/656965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/65/656965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/65/656965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/65/656965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/65/656965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/65/656965/1"}}},"commit":{"parents":[{"commit":"103737ad3cab03a0f6c148b8aef47699928febe4","subject":"update git.openstack.org to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/103737ad3cab03a0f6c148b8aef47699928febe4"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-05-03 15:34:04.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-05-03 15:34:59.000000000","tz":120},"subject":"Add Power and Thermal resources to Chassis","message":"Add Power and Thermal resources to Chassis\n\nAdding the references to Power and Thermal resources to the\nChassis schema.\n\nChange-Id: I041a896f01503ddfac7f343e36b10b9be5318e9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/188a72cee434f4f2cdabd5d71f82e97b28112b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/188a72cee434f4f2cdabd5d71f82e97b28112b2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
