)]}'
{"id":"openstack%2Fsushy~657688","triplet_id":"openstack%2Fsushy~master~Ia1ba7f08c6574c7497ca7464a52f46a5b72bc0cd","project":"openstack/sushy","branch":"master","topic":"task/30593","hashtags":[],"change_id":"Ia1ba7f08c6574c7497ca7464a52f46a5b72bc0cd","subject":"Enhance Storage models to support RAID config","status":"MERGED","created":"2019-05-07 20:57:54.000000000","updated":"2019-06-17 14:27:46.000000000","submitted":"2019-06-17 14:27:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657688-1560781666962-b1f8f8f4","meta_rev_id":"838c98caa3205095677f27ddf641a5539c0db338","_number":657688,"virtual_id_number":657688,"owner":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-04 21:22:57.000000000","_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":2,"date":"2019-06-17 14:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2019-06-17 13:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-14 08:40:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":1,"date":"2019-06-17 13:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 23:07:13.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2019-06-03 08:57:46.000000000","updated_by":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"reviewer":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"state":"REVIEWER"},{"updated":"2019-06-14 08:40:25.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-17 13:30:45.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-06-17 14:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0d42f8e879abd5c98420e1635eeadbd9a3aa94c","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-07 20:57:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b25595c76f8c9c6236239c4d0077b34c093bf392","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 21:15:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/88/657688/1/check/openstack-tox-cover/477bd5e/cover/ : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657688/1/check/openstack-tox-lower-constraints/62ea79b/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/88/657688/1/check/openstack-tox-pep8/a6a469b/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/88/657688/1/check/openstack-tox-py27/1cdb196/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/88/657688/1/check/openstack-tox-py36/93923da/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/88/657688/1/check/openstack-tox-docs/b55c024/html/ : SUCCESS in 6m 25s\n- build-openstack-releasenotes http://logs.openstack.org/88/657688/1/check/build-openstack-releasenotes/5512b07/html/ : SUCCESS in 6m 15s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/88/657688/1/check/sushy-tempest-ironic-partition-redfish-src/f078598/ : FAILURE in 16m 18s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/88/657688/1/check/sushy-tempest-ironic-partition-redfish-src-python2/619d079/ : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"97821404efced31cc9e6749e34f8f638004c940d","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-07 21:28:37.000000000","message":"Patch Set 1:\n\nZuul sushy-tempest-ironic-partition-redfish-src* failuers due to:\n\n2019-05-07 21:13:51.306800 | controller | WARNING: Retrying (Retry(total\u003d0, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)) after connection broken by \u0027NewConnectionError(\u0027\u003cpip._vendor.urllib3.connection.VerifiedHTTPSConnection object at 0x7faeb521e438\u003e: Failed to establish a new connection: [Errno 111] Connection refused\u0027,)\u0027: /openstack/requirements/raw/branch/master/upper-constraints.txt\n\n2019-05-07 21:13:51.307082 | controller | ERROR: Could not install packages due to an EnvironmentError: HTTPSConnectionPool(host\u003d\u0027opendev.org\u0027, port\u003d443): Max retries exceeded with url: /openstack/requirements/raw/branch/master/upper-constraints.txt (Caused by NewConnectionError(\u0027\u003cpip._vendor.urllib3.connection.VerifiedHTTPSConnection object at 0x7faeb521e550\u003e: Failed to establish a new connection: [Errno 111] Connection refused\u0027,))","accounts_in_message":[],"_revision_number":1},{"id":"e4f5dc162c6343355d86f2d8d579618a7cc0ebac","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-10 20:02:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6368b2b7df4f7d3642c45fc53184ce2871e77442","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-10 20:05:40.000000000","message":"Patch Set 2:\n\nPatch set 2 adds delete_volume() method that I forgot to add in initial patch set.","accounts_in_message":[],"_revision_number":2},{"id":"e560918c52307c19eb708e0dc117c37b48c87a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 20:51:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/657688/2/check/openstack-tox-cover/457bf81/cover/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657688/2/check/openstack-tox-lower-constraints/7b5125d/ : SUCCESS in 6m 57s\n- openstack-tox-pep8 http://logs.openstack.org/88/657688/2/check/openstack-tox-pep8/88ea034/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/88/657688/2/check/openstack-tox-py27/6906202/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/88/657688/2/check/openstack-tox-py36/f368cfb/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/88/657688/2/check/openstack-tox-docs/a8f4c3e/html/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/88/657688/2/check/build-openstack-releasenotes/ad682b5/html/ : SUCCESS in 4m 02s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/88/657688/2/check/sushy-tempest-ironic-partition-redfish-src/2d43006/ : SUCCESS in 42m 57s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/88/657688/2/check/sushy-tempest-ironic-partition-redfish-src-python2/d75d2d4/ : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4a6c1a82b94ccb3b10c6c9d427c85e91e5365f67","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-15 18:27:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"51e6668c5e7d33f02e42185df8b013ffc25dc0e9","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-15 18:30:19.000000000","message":"Patch Set 3:\n\nThis patch set adds OperationApplyTimeSupportField to the Volume class.","accounts_in_message":[],"_revision_number":3},{"id":"d7da0686ff12f29f02ab526ace4ad0718c7c8e83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:23:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/657688/3/check/openstack-tox-cover/b57e3ac/cover/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657688/3/check/openstack-tox-lower-constraints/cc36d86/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/88/657688/3/check/openstack-tox-pep8/d6360ee/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/88/657688/3/check/openstack-tox-py27/bba2ff7/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/88/657688/3/check/openstack-tox-py36/523adb9/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/88/657688/3/check/openstack-tox-docs/49d3523/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/88/657688/3/check/build-openstack-releasenotes/874160e/html/ : SUCCESS in 4m 16s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/88/657688/3/check/sushy-tempest-ironic-partition-redfish-src/196209f/ : SUCCESS in 42m 42s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/88/657688/3/check/sushy-tempest-ironic-partition-redfish-src-python2/50ff584/ : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":3},{"id":"99061c3a3c2a1dcecf0aa0fb07fee142e21e9e18","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-05-30 20:41:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"99bb92c6d8272b0f89d27c3f987db05439c7f0c2","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-05-30 21:20:49.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nHi, Julia. Thanks for the review comments. I replied to your comments inline. Let me know if you still have questions or concerns.","accounts_in_message":[],"_revision_number":3},{"id":"44395c9b54958a4febd5785c6d7824947ca8913d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-06-03 08:57:46.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1d94b71d1ec6339484d8fafdc00825906a486bb1","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-06-03 14:11:24.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks Bill!","accounts_in_message":[],"_revision_number":3},{"id":"2b0563bfc2b32cab895056d416b39ae78c6d37fa","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-03 23:42:29.000000000","message":"Patch Set 3:\n\n(5 comments)\n\nThanks for the review comments Julia and Ilya.\n\nIlya - I responded to your comments. Please have a look, especially at the OData-Version header comment in volume.py.","accounts_in_message":[],"_revision_number":3},{"id":"51671ac3abd4a57429b37580197aad53ff9069cf","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-04 13:30:10.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nIlya - I added one new comment re: the Accept header in volume.py.","accounts_in_message":[],"_revision_number":3},{"id":"b00d99828688344da5b9dbe9ba77eebdfa9f6cd8","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-04 21:15:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8bf8464a1ebdf0e8b1be2df5343dcd7ca6477608","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-04 21:22:57.000000000","message":"Patch Set 4:\n\nFor HTTP errors, get the most severe extended info message.\n\nIn the volume.py POST and DELETE requests, removed unneeded \u0027Accept\u0027 and \u0027Content-Type\u0027 headers. Also removed the \u0027OData-Version\u0027 headers here, but will create a new task to add this recommended header in connector.py for all Redfish requests.","accounts_in_message":[],"_revision_number":4},{"id":"9d21fba9b89e6efc31ee625b64be1f3eef1e53bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 22:07:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/88/657688/4/check/openstack-tox-cover/c306df8/cover/ : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657688/4/check/openstack-tox-lower-constraints/08a9b91/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/88/657688/4/check/openstack-tox-pep8/2e4cbd1/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/88/657688/4/check/openstack-tox-py27/9a76687/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/88/657688/4/check/openstack-tox-py36/a85bab0/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/88/657688/4/check/openstack-tox-docs/24da128/html/ : SUCCESS in 5m 46s\n- build-openstack-releasenotes http://logs.openstack.org/88/657688/4/check/build-openstack-releasenotes/e728358/html/ : SUCCESS in 5m 20s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/88/657688/4/check/sushy-tempest-ironic-partition-redfish-src/0262ee6/ : SUCCESS in 43m 59s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/88/657688/4/check/sushy-tempest-ironic-partition-redfish-src-python2/2a50311/ : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":4},{"id":"259624a21c62c114d4b7e2c45ff880c0765dfb54","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-14 08:40:25.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d97f5945c1098823e62d9cf2b43ef35e0ea830bd","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-06-17 13:30:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8240d2281439eaf55a0e9b678327f39ec6b54b15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 13:30:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8ac42b77cff0f668e9dc6c54a43dc5d34663122a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:27:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/88/657688/4/gate/openstack-tox-lower-constraints/b928ba4/ : SUCCESS in 5m 33s\n- openstack-tox-pep8 http://logs.openstack.org/88/657688/4/gate/openstack-tox-pep8/db97ab7/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/88/657688/4/gate/openstack-tox-py27/b1095d5/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/88/657688/4/gate/openstack-tox-py36/4767fcd/ : SUCCESS in 5m 27s\n- openstack-tox-py37 http://logs.openstack.org/88/657688/4/gate/openstack-tox-py37/63652ee/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/88/657688/4/gate/openstack-tox-docs/b318c51/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/88/657688/4/gate/build-openstack-releasenotes/ce79ec8/html/ : SUCCESS in 4m 34s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/88/657688/4/gate/sushy-tempest-ironic-partition-redfish-src/45fdb9d/ : SUCCESS in 55m 07s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/88/657688/4/gate/sushy-tempest-ironic-partition-redfish-src-python2/631c6aa/ : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":4},{"id":"b70c429f9a7d467d72f1b46e694f06eb824a1e0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:27:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"343f7a6c6370cade2e05f57458497cc16681d858","revisions":{"0b95e0e8df538c01d517680415d12171fcfa25de":{"kind":"REWORK","_number":1,"created":"2019-05-07 20:57:54.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/88/657688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/88/657688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/88/657688/1"}}},"commit":{"parents":[{"commit":"363008829a56cceb40e938198b11ae9d756a4aa4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/363008829a56cceb40e938198b11ae9d756a4aa4"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-07 20:50:45.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-07 20:50:45.000000000","tz":-300},"subject":"Enhance Storage models to support RAID config","message":"Enhance Storage models to support RAID config\n\nThe Storage, Volume, and Drive resource models are enhanced to enable\nthe implementation of RAID configuration management.\n\nChange-Id: Ia1ba7f08c6574c7497ca7464a52f46a5b72bc0cd\nStory: 2003514\nTask: 30593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0b95e0e8df538c01d517680415d12171fcfa25de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0b95e0e8df538c01d517680415d12171fcfa25de"}]},"branch":"refs/heads/master"},"b20baf406bcf89b3ee7a1bee314053a2c3fdf6de":{"kind":"REWORK","_number":2,"created":"2019-05-10 20:02:47.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/88/657688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/88/657688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/88/657688/2"}}},"commit":{"parents":[{"commit":"363008829a56cceb40e938198b11ae9d756a4aa4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/363008829a56cceb40e938198b11ae9d756a4aa4"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-07 20:50:45.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-10 19:58:07.000000000","tz":-300},"subject":"Enhance Storage models to support RAID config","message":"Enhance Storage models to support RAID config\n\nThe Storage, Volume, and Drive resource models are enhanced to enable\nthe implementation of RAID configuration management.\n\nChange-Id: Ia1ba7f08c6574c7497ca7464a52f46a5b72bc0cd\nStory: 2003514\nTask: 30593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b20baf406bcf89b3ee7a1bee314053a2c3fdf6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b20baf406bcf89b3ee7a1bee314053a2c3fdf6de"}]},"branch":"refs/heads/master"},"dafbfe317ecb47e6b3b9e76c201e3e277ebe4b04":{"kind":"REWORK","_number":3,"created":"2019-05-15 18:27:16.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/88/657688/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/88/657688/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/88/657688/3"}}},"commit":{"parents":[{"commit":"363008829a56cceb40e938198b11ae9d756a4aa4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/363008829a56cceb40e938198b11ae9d756a4aa4"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-07 20:50:45.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-15 18:24:35.000000000","tz":-300},"subject":"Enhance Storage models to support RAID config","message":"Enhance Storage models to support RAID config\n\nThe Storage, Volume, and Drive resource models are enhanced to enable\nthe implementation of RAID configuration management.\n\nChange-Id: Ia1ba7f08c6574c7497ca7464a52f46a5b72bc0cd\nStory: 2003514\nTask: 30593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dafbfe317ecb47e6b3b9e76c201e3e277ebe4b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dafbfe317ecb47e6b3b9e76c201e3e277ebe4b04"}]},"branch":"refs/heads/master"},"343f7a6c6370cade2e05f57458497cc16681d858":{"kind":"REWORK","_number":4,"created":"2019-06-04 21:15:27.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/88/657688/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/88/657688/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/88/657688/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/88/657688/4"}}},"commit":{"parents":[{"commit":"5a5550d1642337c52956b35da3eca1235efc3749","subject":"Merge \"Add @Redfish.Settings update status and expose it for BIOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5a5550d1642337c52956b35da3eca1235efc3749"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-05-07 20:50:45.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-06-04 21:06:11.000000000","tz":-300},"subject":"Enhance Storage models to support RAID config","message":"Enhance Storage models to support RAID config\n\nThe Storage, Volume, and Drive resource models are enhanced to enable\nthe implementation of RAID configuration management.\n\nChange-Id: Ia1ba7f08c6574c7497ca7464a52f46a5b72bc0cd\nStory: 2003514\nTask: 30593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/343f7a6c6370cade2e05f57458497cc16681d858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/343f7a6c6370cade2e05f57458497cc16681d858"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
