)]}'
{"id":"openstack%2Fsushy~663799","triplet_id":"openstack%2Fsushy~master~Ia74dadf8f383566e00f2ed092f610a8070730aa2","project":"openstack/sushy","branch":"master","topic":"task/33600","hashtags":[],"change_id":"Ia74dadf8f383566e00f2ed092f610a8070730aa2","subject":"Include OData-Version header in Redfish requests","status":"MERGED","created":"2019-06-06 23:44:46.000000000","updated":"2019-06-13 08:31:45.000000000","submitted":"2019-06-13 08:31:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663799-1560414705923-d35274ac","meta_rev_id":"5440998fa2c33ac9667ff2874dbea7a742029a4c","_number":663799,"virtual_id_number":663799,"owner":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-12 14:57:36.000000000","_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2019-06-13 08:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":2,"date":"2019-06-13 03:35:26.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-13 07:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 07:37:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 03:35:26.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-06-13 07:37:50.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-13 08:31:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7e3ce00706b3d39031224279b388b59f4268dbb","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-06 23:44:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42bbe0f37c0aa661ef4a342021f885e6cac8b491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:42:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/99/663799/1/check/openstack-tox-cover/0c023e2/cover/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/663799/1/check/openstack-tox-lower-constraints/62c9f6a/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 http://logs.openstack.org/99/663799/1/check/openstack-tox-pep8/dfa71be/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/99/663799/1/check/openstack-tox-py27/df12c41/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/99/663799/1/check/openstack-tox-py36/0dd229e/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/99/663799/1/check/openstack-tox-docs/b583518/html/ : SUCCESS in 4m 57s\n- build-openstack-releasenotes http://logs.openstack.org/99/663799/1/check/build-openstack-releasenotes/7579676/html/ : SUCCESS in 8m 28s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/99/663799/1/check/sushy-tempest-ironic-partition-redfish-src/1ce290e/ : SUCCESS in 41m 16s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/99/663799/1/check/sushy-tempest-ironic-partition-redfish-src-python2/5f6fff3/ : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a60d4413c54b68b0d3140a1aa549dd8a51a31df6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-07 11:49:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"94fff356098fc850ce4aeccb8c3b7466540067ad","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-07 23:51:13.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nDmitry: Thanks for the review comments. I\u0027ll research the Redfish URI checking some more.","accounts_in_message":[],"_revision_number":1},{"id":"03561d4bc9e21a111b5f70232d57ec9f0829d786","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-10 16:30:47.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nFollow-up comments after a bit more research...","accounts_in_message":[],"_revision_number":1},{"id":"c676ac8aab8419f6ed2989a19d7941f5372c3a52","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-10 17:24:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f7ec92655310d4bf7a3087d9dde513fcda83169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:30:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/99/663799/2/check/openstack-tox-cover/d661d72/cover/ : SUCCESS in 8m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/663799/2/check/openstack-tox-lower-constraints/9e00b51/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/99/663799/2/check/openstack-tox-pep8/d9c06da/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/99/663799/2/check/openstack-tox-py27/052fc36/ : SUCCESS in 5m 19s\n- openstack-tox-py36 http://logs.openstack.org/99/663799/2/check/openstack-tox-py36/81b7ea5/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/99/663799/2/check/openstack-tox-docs/e6eadc7/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/99/663799/2/check/build-openstack-releasenotes/9c47fde/html/ : SUCCESS in 4m 25s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/99/663799/2/check/sushy-tempest-ironic-partition-redfish-src/b98d2aa/ : SUCCESS in 42m 23s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/99/663799/2/check/sushy-tempest-ironic-partition-redfish-src-python2/b9ac740/ : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8499a13940b72c852964d02a0e6f262d59f536d0","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-06-12 06:53:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nThank you Bill. Looks good, one query inline.","accounts_in_message":[],"_revision_number":2},{"id":"54b271aebb2da8daa1c10ed9100719b84a120a81","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-06-12 14:57:36.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nShivanand - thanks for the review! I replied to your question inline.","accounts_in_message":[],"_revision_number":2},{"id":"b3ff2ee3dbab360cab8080f24ecdcae76ddc1cb5","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-06-13 03:35:26.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e (1 comment)\n \u003e \n \u003e Shivanand - thanks for the review! I replied to your question\n \u003e inline.\n\nThank you for the clarification.","accounts_in_message":[],"_revision_number":2},{"id":"032b75276b00e2422fd301a295c80282bdcc8dc1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-13 07:37:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb802507150364d9fe638b9b10f59fb68b4debfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 07:38:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6798d79ec458037cf3cb74521a6ebddf20c0b38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 08:31:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5ff87826ea2088d1a45e7e6e9db5ad674630aa97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 08:31:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/99/663799/2/gate/openstack-tox-lower-constraints/7d2ac94/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/99/663799/2/gate/openstack-tox-pep8/ef79f5e/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/99/663799/2/gate/openstack-tox-py27/b71e673/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/99/663799/2/gate/openstack-tox-py36/8155627/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/99/663799/2/gate/openstack-tox-py37/3e329c4/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/99/663799/2/gate/openstack-tox-docs/507387d/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/99/663799/2/gate/build-openstack-releasenotes/7ff30e6/html/ : SUCCESS in 4m 59s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/99/663799/2/gate/sushy-tempest-ironic-partition-redfish-src/d2f9082/ : SUCCESS in 46m 09s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/99/663799/2/gate/sushy-tempest-ironic-partition-redfish-src-python2/2fde268/ : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6a167c612f806473c6459c849aed8946f2a177f","revisions":{"c7e3ed73e7add21fe45f62540306af73cb8d6049":{"kind":"REWORK","_number":1,"created":"2019-06-06 23:44:46.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/99/663799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/99/663799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/663799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/663799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/663799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/99/663799/1"}}},"commit":{"parents":[{"commit":"6f4539d1b00d31b6f831fc8e23278261c0fdd55e","subject":"Use collections.abc instead of collections when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6f4539d1b00d31b6f831fc8e23278261c0fdd55e"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-06-06 23:42:40.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-06-06 23:42:40.000000000","tz":-300},"subject":"Include OData-Version header in Redfish requests","message":"Include OData-Version header in Redfish requests\n\nIt is recommeded by the Redfish spec to send the OData-Version header\nin requests. This improves interoperability.\n\nThe logic to add the OData-Version header is in the _op() method of\nthe Connector class so that it only needs to be done in one place.\n\nChange-Id: Ia74dadf8f383566e00f2ed092f610a8070730aa2\nStory: 2003514\nTask: 33600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c7e3ed73e7add21fe45f62540306af73cb8d6049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c7e3ed73e7add21fe45f62540306af73cb8d6049"}]},"branch":"refs/heads/master"},"f6a167c612f806473c6459c849aed8946f2a177f":{"kind":"REWORK","_number":2,"created":"2019-06-10 17:24:02.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/99/663799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/99/663799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/663799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/663799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/663799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/99/663799/2"}}},"commit":{"parents":[{"commit":"6f4539d1b00d31b6f831fc8e23278261c0fdd55e","subject":"Use collections.abc instead of collections when available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6f4539d1b00d31b6f831fc8e23278261c0fdd55e"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-06-06 23:42:40.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-06-10 17:22:58.000000000","tz":-300},"subject":"Include OData-Version header in Redfish requests","message":"Include OData-Version header in Redfish requests\n\nIt is recommeded by the Redfish spec to send the OData-Version header\nin requests. This improves interoperability.\n\nThe logic to add the OData-Version header is in the _op() method of\nthe Connector class so that it only needs to be done in one place.\n\nChange-Id: Ia74dadf8f383566e00f2ed092f610a8070730aa2\nStory: 2003514\nTask: 33600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f6a167c612f806473c6459c849aed8946f2a177f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f6a167c612f806473c6459c849aed8946f2a177f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
