)]}'
{"id":"openstack%2Fsushy~665342","triplet_id":"openstack%2Fsushy~master~Iadbfcb0ab746ac0a64abc85ae7ea2f628405f6fb","project":"openstack/sushy","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Iadbfcb0ab746ac0a64abc85ae7ea2f628405f6fb","subject":"Build pdf doc","status":"MERGED","created":"2019-06-14 09:40:29.000000000","updated":"2019-09-05 11:44:28.000000000","submitted":"2019-09-05 11:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665342-1567683776368-07ca9f8f","meta_rev_id":"51a126ec76abf3a011d75d52048f9b1242068a73","_number":665342,"virtual_id_number":665342,"owner":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-09-05 11:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2019-09-05 10:52:39.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-05 10:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2019-09-05 10:52:39.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 09:48:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-09-05 10:51:59.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-09-05 10:52:39.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-09-05 11:42:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d95b9fd0b95c1a514ffc4e2b7c57cbba1e2ce8ee","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-06-14 09:40:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4610fbbde432f3edb15c90861c2a9dc7a73c077","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-14 09:48:48.000000000","message":"Patch Set 1:\n\ntested locally and looks good :)","accounts_in_message":[],"_revision_number":1},{"id":"e14aa198baba4676e8ff08a0b67da41c7d2cb15a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 10:36:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/42/665342/1/check/requirements-check/9947fdf/ : SUCCESS in 5m 32s\n- openstack-tox-cover http://logs.openstack.org/42/665342/1/check/openstack-tox-cover/b063b8d/cover/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/665342/1/check/openstack-tox-lower-constraints/7210c69/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/42/665342/1/check/openstack-tox-pep8/d932295/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/42/665342/1/check/openstack-tox-py27/d18cc18/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/42/665342/1/check/openstack-tox-py36/0312606/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/42/665342/1/check/openstack-tox-py37/e286eba/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/42/665342/1/check/openstack-tox-docs/0bb5039/html/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/42/665342/1/check/build-openstack-releasenotes/7a75723/html/ : SUCCESS in 4m 39s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/42/665342/1/check/sushy-tempest-ironic-partition-redfish-src/66c4e6f/ : SUCCESS in 49m 42s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/42/665342/1/check/sushy-tempest-ironic-partition-redfish-src-python2/5a3f1bb/ : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":1},{"id":"db379095efa2e47cf36e9b4b7d4f0fbce5707d24","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-09-05 06:25:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e226536a3e9c436241f78affbd268e05495068f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:45:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b6b31ea67434db499937996bb72621f : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b3d70e913374105bf7c30f5659e42c2 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/857ed194bd1b4a65802c01d1567e1bb2 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d09fc5d652432481df2252bf61c082 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1dd0487c7c3240e5a572dde416d8a870 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81671dbb07d845b18c34f7d50551c6b5 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31661e4675a34fa5a46a334baf3399b9 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da955077d41f4931913bf0834363d9e9 : FAILURE in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4a9dcf80a184117a2a558b94da813b9 : SUCCESS in 4m 10s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9484e058d3b645d29450d80fb787c69d : SUCCESS in 48m 49s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/23bbf7c73a804fd3971d56185ec4e8df : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":2},{"id":"2c3edfa5acbbbf165ad957ab4d8b43eb4471f0c4","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-09-05 09:24:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba0eae51d2a8ddfe095fa8020098c57796bb91ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 10:20:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac55e17ff4ce4911bbc1c87049eb1b34 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dee88b1084b4afa9748a3c6991c7da0 : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b45cd6d9a24644e2ab72f2c648a97ef5 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d2e7313d7c4d9183df6a00deb93e7b : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/937d67048e46435692d923623b324fb5 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7df4aa33859d4d5e8b9a01213debd6e5 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e60dba61bcb49d088b8d38d3341c304 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6435ce3d2ec4f3e911ad2c72749286a : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e26c6bb1c6b64a5689f149b04e239e9f : SUCCESS in 4m 21s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/289a666afcdd4b88a6eb795124e29ed0 : SUCCESS in 49m 33s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/5b26583d36cc468081dc6234e13faf86 : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b14285671c23a44fc4aedc686fe3a4a076605b8c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-09-05 10:51:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a013f3898eddf51ece29f40254e058e7ac9f38e8","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-09-05 10:52:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThank you Kaifeng.","accounts_in_message":[],"_revision_number":3},{"id":"9812b1cdcd747f2ecd09396950c5c5e1e0bc7e1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 10:52:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eeb117d0181ed2cbbed23c5d6aaf87e7cc834c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:42:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9491877d7d643b994b64786a6dd7610 : SUCCESS in 2m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8801a04151e4273a82bc05fc983f09d : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee205dd7286443d8beabbd7beb23661 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdda73aeb5844182a1a405661bbf1e11 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eda6992d0c4e4ad9b243d7cc902e6edd : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92cda12ac249425283c41fa460b11ddf : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/230b64bcd6de4bb1b15caccd78ea00f8 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcf34289398347c7a57053ab5eb9c2c8 : SUCCESS in 4m 01s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/1da55f6901b5495da0594d827b95d5e9 : SUCCESS in 43m 35s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/42618c4188f44efd974a32fb446df016 : SUCCESS in 43m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1b898d18b3d90281f3b5d9075f979f570d8d4b06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:42:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e408f34ef474ddc8dd102cc18f34f48582b7da2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:44:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de73aa16d03b4f1bbb2ddaa03640eb90 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae2bafb37fe5bf2762c2182928f02b02f000eb9f","revisions":{"cc618a9fda60e14f82e4c903bf54b19cfcce8ca6":{"kind":"REWORK","_number":1,"created":"2019-06-14 09:40:29.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/42/665342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/42/665342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/42/665342/1"}}},"commit":{"parents":[{"commit":"c209cb805adbb068e3a068816d67e17994bbfb8d","subject":"Merge \"Include OData-Version header in Redfish requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c209cb805adbb068e3a068816d67e17994bbfb8d"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-14 09:34:30.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-14 09:34:30.000000000","tz":480},"subject":"[PoC] Enable creating pdf doc","message":"[PoC] Enable creating pdf doc\n\nSee https://review.opendev.org/#/c/665285 for more.\n\nChange-Id: Iadbfcb0ab746ac0a64abc85ae7ea2f628405f6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cc618a9fda60e14f82e4c903bf54b19cfcce8ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cc618a9fda60e14f82e4c903bf54b19cfcce8ca6"}]},"branch":"refs/heads/master"},"403a0f3723f4195bc8e68ceb8dbfeba390cfb2fd":{"kind":"REWORK","_number":2,"created":"2019-09-05 06:25:27.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/42/665342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/42/665342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/42/665342/2"}}},"commit":{"parents":[{"commit":"c209cb805adbb068e3a068816d67e17994bbfb8d","subject":"Merge \"Include OData-Version header in Redfish requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c209cb805adbb068e3a068816d67e17994bbfb8d"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-14 09:34:30.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-09-05 06:23:55.000000000","tz":480},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project could produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is recommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Iadbfcb0ab746ac0a64abc85ae7ea2f628405f6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/403a0f3723f4195bc8e68ceb8dbfeba390cfb2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/403a0f3723f4195bc8e68ceb8dbfeba390cfb2fd"}]},"branch":"refs/heads/master"},"ae2bafb37fe5bf2762c2182928f02b02f000eb9f":{"kind":"REWORK","_number":3,"created":"2019-09-05 09:24:43.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/42/665342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/42/665342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/42/665342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/42/665342/3"}}},"commit":{"parents":[{"commit":"c209cb805adbb068e3a068816d67e17994bbfb8d","subject":"Merge \"Include OData-Version header in Redfish requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c209cb805adbb068e3a068816d67e17994bbfb8d"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-14 09:34:30.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-09-05 09:23:31.000000000","tz":480},"subject":"Build pdf doc","message":"Build pdf doc\n\nThe is one of community goals that each project could produce a\nsingle PDF file. The pdf should be in the output of openstack-tox-docs\njob.\n\nTeX packages are required to build PDF locally, following is recommended:\n\n* inkscape\n* texlive-latex-base\n* texlive-latex-extra\n* texlive-fonts-recommended\n\nMore about the goal:\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\nhttps://etherpad.openstack.org/p/train-pdf-support-goal\nhttps://etherpad.openstack.org/p/pdf-goal-train-common-problems\n\nChange-Id: Iadbfcb0ab746ac0a64abc85ae7ea2f628405f6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ae2bafb37fe5bf2762c2182928f02b02f000eb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ae2bafb37fe5bf2762c2182928f02b02f000eb9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
