)]}'
{"id":"openstack%2Fsushy~667192","triplet_id":"openstack%2Fsushy~master~I86d30b836deca1994a01aac320ae815d28846a53","project":"openstack/sushy","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I86d30b836deca1994a01aac320ae815d28846a53","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2019-06-24 19:13:01.000000000","updated":"2019-07-08 14:48:36.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8312d3aeaf5c2020052f9ac173c576f54481fc99","_number":667192,"virtual_id_number":667192,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-03 14:13:46.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2019-06-26 19:35:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2019-07-05 14:32:22.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":-1,"date":"2019-07-06 19:33:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 11:25:17.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-26 19:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 14:13:46.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-07-06 19:33:53.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"f41bb1079c57584035982ce6ecdc5749ba567e49","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-24 19:13:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaea5ba3be9c01b00b40031005367e7cd61a4c89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 20:02:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/92/667192/1/check/openstack-tox-cover/e546497/cover/ : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/667192/1/check/openstack-tox-lower-constraints/786b4b7/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/92/667192/1/check/openstack-tox-pep8/8f76e44/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/92/667192/1/check/openstack-tox-py27/2da37a9/ : SUCCESS in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/92/667192/1/check/openstack-tox-py36/9e964ec/ : SUCCESS in 3m 00s\n- openstack-tox-py37 http://logs.openstack.org/92/667192/1/check/openstack-tox-py37/e1a498f/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/92/667192/1/check/openstack-tox-docs/3fab05f/html/ : SUCCESS in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/92/667192/1/check/build-openstack-releasenotes/d7f626c/html/ : SUCCESS in 3m 19s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/667192/1/check/sushy-tempest-ironic-partition-redfish-src/432d751/ : SUCCESS in 42m 36s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/92/667192/1/check/sushy-tempest-ironic-partition-redfish-src-python2/e3c2cea/ : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":1},{"id":"581b6650561ce8b9d4e83848b7f379d3adfc5aa3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-25 11:25:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0591b74dc1414fd1e3b190a67fa1053f26c2e4b8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-26 08:42:52.000000000","message":"Patch Set 1: Code-Review-1\n\nDitto as other patches","accounts_in_message":[],"_revision_number":1},{"id":"74470779fc0b708a98cdcd8f15a252b3ff92c389","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-26 18:40:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"36da1677533129bb55dc3ad094d4941dc36d2eb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:35:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/92/667192/2/check/openstack-tox-cover/ac22c93/cover/ : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/667192/2/check/openstack-tox-lower-constraints/02bd5f5/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/92/667192/2/check/openstack-tox-pep8/a3d9d6c/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/92/667192/2/check/openstack-tox-py27/dba18f8/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/92/667192/2/check/openstack-tox-py36/bfb2475/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/92/667192/2/check/openstack-tox-py37/b22ca12/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/92/667192/2/check/openstack-tox-docs/077650c/html/ : SUCCESS in 3m 37s\n- build-openstack-releasenotes http://logs.openstack.org/92/667192/2/check/build-openstack-releasenotes/c896b80/html/ : SUCCESS in 3m 17s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/92/667192/2/check/sushy-tempest-ironic-partition-redfish-src/b46896a/ : SUCCESS in 46m 34s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/92/667192/2/check/sushy-tempest-ironic-partition-redfish-src-python2/3c7ea53/ : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a6fd188c579ca5daccaf13329592a8b89d2e90d1","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-06-26 19:59:34.000000000","message":"Patch Set 2:\n\n@Dmitiry feel free to ingore this patch update. I didn\u0027t understand \"ditto as other patches\" until later. I know you want to stick with py3. Note though, I think that defaults to whatever py3 is default on the developer\u0027s system, so the developer could be running py34 and gate will be running py36/py37.","accounts_in_message":[],"_revision_number":2},{"id":"cba312d98515ac7cff09e7a843524658a4adced9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-27 10:25:59.000000000","message":"Patch Set 2:\n\n\u003e Note though, I think that defaults to whatever py3 is default on the developer\u0027s system, so the developer could be running py34 and gate will be running py36/py37.\n\nExactly. And we cannot change that. I\u0027d like to avoid the situation when \u0027tox\u0027 fails because a developer has python 3.6 only.","accounts_in_message":[],"_revision_number":2},{"id":"f64263b7df5f9d622888bc78ab456937b95d3358","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-02 18:49:02.000000000","message":"Patch Set 2:\n\nFair enough and good point, but time to upgrade if you don\u0027t have py37 :)","accounts_in_message":[],"_revision_number":2},{"id":"c5b8985c436709537d71e1e7bbca0e9cd484eb61","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-07-03 14:13:46.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Fair enough and good point, but time to upgrade if you don\u0027t have\n \u003e py37 :)\n\nTrue, but depending on the distribution, that is far harder than one would think. The point of the gate is that it does the right thing in the end for us. It tests and blocks on the versions that we need it to. If a developer is off by one version for running unit tests than I think that is okay as long as the gate still does the right thing and blocks anything not compatible with newer versions.","accounts_in_message":[],"_revision_number":2},{"id":"0a4a07f53328eb0ed94d6c6fb7817c4ad200300c","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 14:31:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70a638449acfb8e1760f23408dc1653895b3d3bd","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 14:32:22.000000000","message":"Patch Set 2:\n\nI should say \"moved so far\" as the python3 goal for train is still progressing.","accounts_in_message":[],"_revision_number":2},{"id":"bf06448910521edca44e8b2b644f9d8a5b0e736b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-06 19:33:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da9d3dc5d4d440f8f0bee7aa64d70fa439480cbc","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-08 14:48:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4644abcc002800fde633055aa42d98b881266e5b","revisions":{"a932f1913bc1f7a352823dc850a6d1f0b902248d":{"kind":"REWORK","_number":1,"created":"2019-06-24 19:13:01.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/92/667192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/92/667192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/92/667192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/92/667192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/92/667192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/92/667192/1"}}},"commit":{"parents":[{"commit":"640c567bae6c11fb4a8f92ec591066ad400e80cb","subject":"Merge \"Make Manager-\u003eActions field optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/640c567bae6c11fb4a8f92ec591066ad400e80cb"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 18:52:38.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 19:13:00.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I86d30b836deca1994a01aac320ae815d28846a53\nStory: #2005924\nTask: #34213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a932f1913bc1f7a352823dc850a6d1f0b902248d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a932f1913bc1f7a352823dc850a6d1f0b902248d"}]},"branch":"refs/heads/master"},"4644abcc002800fde633055aa42d98b881266e5b":{"kind":"REWORK","_number":2,"created":"2019-06-26 18:40:41.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/92/667192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/92/667192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/92/667192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/92/667192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/92/667192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/92/667192/2"}}},"commit":{"parents":[{"commit":"640c567bae6c11fb4a8f92ec591066ad400e80cb","subject":"Merge \"Make Manager-\u003eActions field optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/640c567bae6c11fb4a8f92ec591066ad400e80cb"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-24 18:52:38.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-06-26 18:40:37.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I86d30b836deca1994a01aac320ae815d28846a53\nStory: #2005924\nTask: #34213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4644abcc002800fde633055aa42d98b881266e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4644abcc002800fde633055aa42d98b881266e5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
