)]}'
{"id":"openstack%2Fsushy~669963","triplet_id":"openstack%2Fsushy~master~I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b","project":"openstack/sushy","branch":"master","topic":"story/2006195","hashtags":[],"change_id":"I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b","subject":"Implements adapter checking","status":"MERGED","created":"2019-07-10 01:50:17.000000000","updated":"2020-01-21 17:09:38.000000000","submitted":"2019-08-14 12:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669963-1565787238741-d4deb22e","meta_rev_id":"e3d83b97481ef87fe56b66e7cceb7d5fafa5c90d","_number":669963,"virtual_id_number":669963,"owner":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-08-14 12:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2019-08-02 21:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2019-08-14 12:32:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-08-14 12:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-08-14 12:04:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 21:10:16.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-14 12:04:01.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-08-14 12:32:17.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-14 12:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4c049cd05cf4e11c2214af0be58c5bc82cdab23","author":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"date":"2019-07-10 01:50:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b250e7f7e0b1b033a0313c504d73e26ea0df4234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 02:28:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/63/669963/1/check/openstack-tox-cover/5afd480/ : FAILURE in 2m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/669963/1/check/openstack-tox-lower-constraints/c72f0c8/ : FAILURE in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/63/669963/1/check/openstack-tox-pep8/2416422/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/63/669963/1/check/openstack-tox-py27/c16a891/ : FAILURE in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/63/669963/1/check/openstack-tox-py36/1d39ea1/ : FAILURE in 2m 41s\n- openstack-tox-py37 http://logs.openstack.org/63/669963/1/check/openstack-tox-py37/49e166b/ : FAILURE in 2m 57s\n- openstack-tox-docs http://logs.openstack.org/63/669963/1/check/openstack-tox-docs/1fd9387/html/ : SUCCESS in 3m 10s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/63/669963/1/check/sushy-tempest-ironic-partition-redfish-src/15a0182/ : FAILURE in 29m 22s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/63/669963/1/check/sushy-tempest-ironic-partition-redfish-src-python2/ef8d160/ : FAILURE in 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"45b03b5c03744cd3a69dcbb4252b4fb1c544bdf9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-10 13:15:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b17dfb397eff772ce44e89d3d58557c4d42db8b9","author":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"date":"2019-07-10 18:01:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f8bc585249b6a0127496b819bc4303caa015b9e9","author":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"date":"2019-07-10 18:03:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1c29bf9481732089314ceadbba6cf9b3ee07a69","author":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"date":"2019-07-10 18:06:52.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"60321a977d80ed00bb5141f081fe6dd17879aa5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:57:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/63/669963/4/check/openstack-tox-cover/12c8293/cover/ : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/669963/4/check/openstack-tox-lower-constraints/c19c5cf/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/63/669963/4/check/openstack-tox-pep8/1731f3e/ : FAILURE in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/63/669963/4/check/openstack-tox-py27/04facd1/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/63/669963/4/check/openstack-tox-py36/bf21fd0/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/63/669963/4/check/openstack-tox-py37/71d2313/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/63/669963/4/check/openstack-tox-docs/789f885/html/ : SUCCESS in 2m 52s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/63/669963/4/check/sushy-tempest-ironic-partition-redfish-src/90d53ea/ : SUCCESS in 43m 01s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/63/669963/4/check/sushy-tempest-ironic-partition-redfish-src-python2/8f903ca/ : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":4},{"id":"186cde51b8575cbc1cdbce9abb8678826dbf8b31","author":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"date":"2019-07-10 19:10:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6f48fe0459c571595a37204b5ba79974e56b7212","author":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"date":"2019-07-10 19:47:43.000000000","message":"Topic changed from story/35777 to story/2006195","accounts_in_message":[],"_revision_number":5},{"id":"18f34ca07fa61c5482f16aad429cdbce94fccbcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:09:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/63/669963/5/check/openstack-tox-cover/6e95f82/cover/ : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/669963/5/check/openstack-tox-lower-constraints/037b192/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/63/669963/5/check/openstack-tox-pep8/378204b/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/63/669963/5/check/openstack-tox-py27/c2cc2c3/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/63/669963/5/check/openstack-tox-py36/2813a99/ : SUCCESS in 3m 38s\n- openstack-tox-py37 http://logs.openstack.org/63/669963/5/check/openstack-tox-py37/5cd5651/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/63/669963/5/check/openstack-tox-docs/a5ca577/html/ : SUCCESS in 3m 02s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/63/669963/5/check/sushy-tempest-ironic-partition-redfish-src/73bf581/ : SUCCESS in 51m 35s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/63/669963/5/check/sushy-tempest-ironic-partition-redfish-src-python2/4214434/ : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":5},{"id":"a546e5fdb45fcc1bdfb8cff478396171735962fc","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-02 21:09:25.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"21f2a627ff035434362e5b5ae5247e70c9f1d273","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-02 21:10:16.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM! I suspect this might be release note worthy, but kind of makes the library a little more permissive which makes sense to me. Thanks!","accounts_in_message":[],"_revision_number":6},{"id":"fd1a50e12638f891badc55cbc080213f532fa0a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 22:02:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/63/669963/6/check/openstack-tox-cover/9bc645f/cover/ : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints https://logs.opendev.org/63/669963/6/check/openstack-tox-lower-constraints/31bf038/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://logs.opendev.org/63/669963/6/check/openstack-tox-pep8/99c8b6d/ : SUCCESS in 3m 51s\n- openstack-tox-py27 https://logs.opendev.org/63/669963/6/check/openstack-tox-py27/5cfa369/ : SUCCESS in 3m 01s\n- openstack-tox-py36 https://logs.opendev.org/63/669963/6/check/openstack-tox-py36/6575744/ : SUCCESS in 3m 39s\n- openstack-tox-py37 https://logs.opendev.org/63/669963/6/check/openstack-tox-py37/e548cfe/ : SUCCESS in 3m 17s\n- openstack-tox-docs https://logs.opendev.org/63/669963/6/check/openstack-tox-docs/a953a4b/html/ : SUCCESS in 3m 25s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/63/669963/6/check/sushy-tempest-ironic-partition-redfish-src/25202b9/ : SUCCESS in 44m 42s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/63/669963/6/check/sushy-tempest-ironic-partition-redfish-src-python2/26762c1/ : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":6},{"id":"9a86bc0216a71c140404c5a245c6a5bb2321af58","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-08-14 12:04:01.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0855f4299271d7892ea05cd3b777f92ac59dc969","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 12:04:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cdb76d7aa475f6ae83ca9459dd3ede4bdd30a395","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 12:32:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6155e177aef7987c0147e26e5a059038a2a4281a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 12:53:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8da862a6c33494cb76e7b3f92ec6d56 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679b59dd466349b1ad0a3f6c5c400de8 : SUCCESS in 2m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b235af17c7949a08ec979fb4e995950 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96b39226df444b829062a4437e55977f : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dba8668534a4ac1aab65566d16647c9 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c0b54ae8c2a4a63b610d6c470bd1403 : SUCCESS in 2m 33s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/7ed51df70f4c47faab7f8287a72de47c : SUCCESS in 47m 37s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/8006525d4f0d4ef0bd2991743d415da4 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":6},{"id":"ddc90ff6068048972fec05618f4b13d7f304efd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 12:53:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"418e130353e820e158b4fee9b29cee8e6358ceae","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-16 12:30:51.000000000","message":"Patch Set 6:\n\nI am sorry for being late to the party, but could you please share an example OEM JSON causing the problem you were solving?","accounts_in_message":[],"_revision_number":6},{"id":"e3d83b97481ef87fe56b66e7cceb7d5fafa5c90d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-21 17:09:38.000000000","message":"Patch Set 6:\n\nSee also: https://review.opendev.org/703652","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"72fd054da0b007e84b81728e7357c2d76c45e666","revisions":{"19de938b304eb476507b1aeefcceca26d9c1bf20":{"kind":"REWORK","_number":1,"created":"2019-07-10 01:50:17.000000000","uploader":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"ref":"refs/changes/63/669963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/669963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/669963/1"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:36:42.000000000","tz":-420},"committer":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:50:08.000000000","tz":-420},"subject":"Adds support for allowing a sushy object to be initialized when a vendor does not implement all base resources. Implements adapter checking Closes Story: 35777","message":"Adds support for allowing a sushy object to be initialized when a vendor does not implement all base resources.\nImplements adapter checking\nCloses Story: 35777\n\nChange-Id: I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19de938b304eb476507b1aeefcceca26d9c1bf20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19de938b304eb476507b1aeefcceca26d9c1bf20"}]},"branch":"refs/heads/master"},"a1eb8e615a88b76a339b47050d229c41ec279243":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-10 18:01:33.000000000","uploader":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"ref":"refs/changes/63/669963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/669963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/669963/2"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:36:42.000000000","tz":-420},"committer":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 18:00:27.000000000","tz":-420},"subject":"Implements adapter checking","message":"Implements adapter checking\n\nAdds support for allowing a sushy object to be initialized when a vendor does not implement all base resources.\nCloses Story: 35777\n\nChange-Id: I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a1eb8e615a88b76a339b47050d229c41ec279243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a1eb8e615a88b76a339b47050d229c41ec279243"}]},"branch":"refs/heads/master"},"3f3d24cd658c8d2f30ea29b64a2ae6a4dcdc2f6d":{"kind":"REWORK","_number":3,"created":"2019-07-10 18:03:09.000000000","uploader":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"ref":"refs/changes/63/669963/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/669963/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/669963/3"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:36:42.000000000","tz":-420},"committer":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 18:03:01.000000000","tz":-420},"subject":"Implements adapter checking","message":"Implements adapter checking\n\nAdds support for allowing a sushy object to be initialized when a vendor does not implement all base resources.\nCloses Story: 35777\n\nChange-Id: I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3f3d24cd658c8d2f30ea29b64a2ae6a4dcdc2f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3f3d24cd658c8d2f30ea29b64a2ae6a4dcdc2f6d"}]},"branch":"refs/heads/master"},"a97010b5c792c007598d75033a84c627adbd6bc0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-07-10 18:06:52.000000000","uploader":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"ref":"refs/changes/63/669963/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/669963/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/669963/4"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:36:42.000000000","tz":-420},"committer":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 18:06:06.000000000","tz":-420},"subject":"Implements adapter checking","message":"Implements adapter checking\n\nAdds support for allowing a sushy object to be initialized when a vendor does not implement all base resources.\nStory: 2006195\nTask: 35777\n\nChange-Id: I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a97010b5c792c007598d75033a84c627adbd6bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a97010b5c792c007598d75033a84c627adbd6bc0"}]},"branch":"refs/heads/master"},"50b3212531da57fa3655aa7f78711097f43361eb":{"kind":"REWORK","_number":5,"created":"2019-07-10 19:10:03.000000000","uploader":{"_account_id":26143,"name":"Yusef Shaban","email":"yshaban@godaddy.com","username":"yshaban"},"ref":"refs/changes/63/669963/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/669963/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/669963/5"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:36:42.000000000","tz":-420},"committer":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 19:09:55.000000000","tz":-420},"subject":"Implements adapter checking","message":"Implements adapter checking\n\nAdds support for allowing a sushy object to be initialized when a vendor does not implement all base resources.\nStory: 2006195\nTask: 35777\n\nChange-Id: I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/50b3212531da57fa3655aa7f78711097f43361eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/50b3212531da57fa3655aa7f78711097f43361eb"}]},"branch":"refs/heads/master"},"72fd054da0b007e84b81728e7357c2d76c45e666":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-08-02 21:09:25.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/63/669963/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/669963/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/669963/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/669963/6"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Yusef Shaban","email":"yshaban@godaddy.com","date":"2019-07-10 01:36:42.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2019-08-02 21:09:23.000000000","tz":0},"subject":"Implements adapter checking","message":"Implements adapter checking\n\nAdds support for allowing a sushy object to be initialized when a\nvendor does not implement all base resources.\n\nStory: 2006195\nTask: 35777\nChange-Id: I59b6553d8d5aa60ad4e79df9507c9fe9fc1dc86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/72fd054da0b007e84b81728e7357c2d76c45e666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/72fd054da0b007e84b81728e7357c2d76c45e666"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
