)]}'
{"id":"openstack%2Fsushy~670360","triplet_id":"openstack%2Fsushy~master~Ice6700806710f37b590f6b879ec38656326b39c6","project":"openstack/sushy","branch":"master","topic":"task/30719","hashtags":[],"change_id":"Ice6700806710f37b590f6b879ec38656326b39c6","subject":"Add Task Monitor support for async operations","status":"MERGED","created":"2019-07-11 18:21:12.000000000","updated":"2020-01-27 15:28:48.000000000","submitted":"2020-01-27 15:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":62,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670360-1580138750693-46a14a57","meta_rev_id":"53e321d4e9ba6699f419d686a300c8d731e5309c","_number":670360,"virtual_id_number":670360,"owner":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"date":"2020-01-20 21:28:43.000000000","_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"date":"2020-01-21 01:35:53.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 15:25:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2020-01-27 13:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2020-01-27 14:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":1,"date":"2020-01-27 14:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 22:43:21.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-07-16 22:44:02.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2019-11-01 11:44:59.000000000","updated_by":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"reviewer":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"state":"REVIEWER"},{"updated":"2020-01-20 21:28:43.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-01-21 01:35:53.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-01-27 13:56:13.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-01-27 14:23:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-01-27 15:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b6347fce42998e2f959d45e7b5cf8cf7072ad40","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-07-11 18:21:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee3b334ab03ebc9414e19434d65af1c1e51a0c74","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-07-11 19:09:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9aa2be7da0b9c7cd37c2e56ad65e5d7bf1fdffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 20:02:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/60/670360/2/check/openstack-tox-cover/805ddd1/cover/ : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/670360/2/check/openstack-tox-lower-constraints/e73a382/ : SUCCESS in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/60/670360/2/check/openstack-tox-pep8/1ac3d36/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/60/670360/2/check/openstack-tox-py27/01e57a9/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/60/670360/2/check/openstack-tox-py36/dfa32fb/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/60/670360/2/check/openstack-tox-py37/be066a8/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/60/670360/2/check/openstack-tox-docs/67fdecf/html/ : SUCCESS in 2m 28s\n- build-openstack-releasenotes http://logs.openstack.org/60/670360/2/check/build-openstack-releasenotes/3ef84db/html/ : SUCCESS in 3m 19s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/670360/2/check/sushy-tempest-ironic-partition-redfish-src/5ece138/ : SUCCESS in 44m 07s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/670360/2/check/sushy-tempest-ironic-partition-redfish-src-python2/a8ae069/ : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ac58dea158225c51a8558534776dbc864d08cfdc","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-18 08:25:55.000000000","message":"Patch Set 2:\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5894ea5b116f384f738c5aa7c4e17555e3d17bf1","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-07-18 23:28:36.000000000","message":"Patch Set 2:\n\n(11 comments)\n\nThanks for the great review comments, Ilya! Updates coming soon in patch set 3.","accounts_in_message":[],"_revision_number":2},{"id":"84e973dcb3d4876c7136e4262d80fe837065902f","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-07-18 23:46:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2e3850bb4393017a9d9ef58cbfa7c6ae514e1fb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 00:35:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/60/670360/3/check/openstack-tox-cover/491ed4c/cover/ : SUCCESS in 2m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/670360/3/check/openstack-tox-lower-constraints/fb708d4/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/60/670360/3/check/openstack-tox-pep8/ecd2361/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/60/670360/3/check/openstack-tox-py27/dcedde1/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/60/670360/3/check/openstack-tox-py36/e9399b0/ : SUCCESS in 2m 43s\n- openstack-tox-py37 http://logs.openstack.org/60/670360/3/check/openstack-tox-py37/0d64fee/ : SUCCESS in 3m 16s\n- openstack-tox-docs http://logs.openstack.org/60/670360/3/check/openstack-tox-docs/9889ee1/html/ : SUCCESS in 2m 42s\n- build-openstack-releasenotes http://logs.openstack.org/60/670360/3/check/build-openstack-releasenotes/c26ab9e/html/ : SUCCESS in 3m 25s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/670360/3/check/sushy-tempest-ironic-partition-redfish-src/7eab167/ : SUCCESS in 47m 45s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/670360/3/check/sushy-tempest-ironic-partition-redfish-src-python2/48e61b4/ : FAILURE in 46m 05s","accounts_in_message":[],"_revision_number":3},{"id":"4cc0d97a069e53c0cbcde5e6a2546b9e1345af49","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-07-19 01:11:07.000000000","message":"Patch Set 3:\n\nI\u0027m not sure, but I\u0027m assuming the Zuul failure is a temporary pypi server error:\n\n  Collecting oslo.concurrency\u003d\u003d\u003d3.29.1 (from -c https://releases.openstack.org/constraints/upper/master (line 24))\n    WARNING: Retrying (Retry(total\u003d4, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)) after connection broken by \u0027NewConnectionError(\u0027\u003cpip._vendor.urllib3.connection.HTTPConnection object at 0x7fc44d2ccb90\u003e: Failed to establish a new connection: [Errno -3] Temporary failure in name resolution\u0027,)\u0027: /pypi/simple/oslo-concurrency/\n    ...\n    WARNING: Retrying (Retry(total\u003d0, connect\u003dNone, read\u003dNone, redirect\u003dNone, status\u003dNone)) after connection broken by \u0027NewConnectionError(\u0027\u003cpip._vendor.urllib3.connection.HTTPConnection object at 0x7fc44d086a50\u003e: Failed to establish a new connection: [Errno -3] Temporary failure in name resolution\u0027,)\u0027: /wheel/ubuntu-18.04-x86_64/oslo-concurrency/\n    ERROR: Could not find a version that satisfies the requirement oslo.concurrency\u003d\u003d\u003d3.29.1 (from -c https://releases.openstack.org/constraints/upper/master (line 24)) (from versions: none)\n  ERROR: No matching distribution found for oslo.concurrency\u003d\u003d\u003d3.29.1 (from -c https://releases.openstack.org/constraints/upper/master (line 24))","accounts_in_message":[],"_revision_number":3},{"id":"fc1c98868f6150a949c2bb55a227b71c25ddf7ac","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-07-21 16:19:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"79012785162a70380bcb287b24a256360e6538a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-21 17:03:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/60/670360/3/check/openstack-tox-cover/952c793/cover/ : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/670360/3/check/openstack-tox-lower-constraints/5623b03/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/60/670360/3/check/openstack-tox-pep8/4a45353/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/60/670360/3/check/openstack-tox-py27/14bdcb3/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/60/670360/3/check/openstack-tox-py36/4f05203/ : SUCCESS in 3m 37s\n- openstack-tox-py37 http://logs.openstack.org/60/670360/3/check/openstack-tox-py37/ec87cbb/ : SUCCESS in 4m 08s\n- openstack-tox-docs http://logs.openstack.org/60/670360/3/check/openstack-tox-docs/e106e65/html/ : SUCCESS in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/60/670360/3/check/build-openstack-releasenotes/35e21f6/html/ : SUCCESS in 3m 26s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/60/670360/3/check/sushy-tempest-ironic-partition-redfish-src/ccdf1f9/ : SUCCESS in 39m 31s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/60/670360/3/check/sushy-tempest-ironic-partition-redfish-src-python2/2709ebd/ : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":3},{"id":"adf4ce74189e265bf964aa8e689962a1a8b74aaa","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-02 09:19:49.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nnitpick: arguably, having task monitor feature in one patch and volume using it could be another...","accounts_in_message":[],"_revision_number":2},{"id":"1a1d85b27bab83f99259a7b10ab1f022360eaeb9","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-08-07 16:21:37.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nIlya - thanks for the review comments. Regarding the comment about doing the volume.py updates in a different patch set: Yes, I did think of that. One of the reasons I did it here was to show how the TaskMonitor would be used in context. Do you feel I do need to split it out or is it OK to leave in this patch?\n\nI\u0027ll fix the retry-after setter in patch set 4.","accounts_in_message":[],"_revision_number":3},{"id":"569fbfac0757d5c5af5d0ad496f551a66dcac76e","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-08-08 16:42:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96b95af4fcbb1b1d3ea6fa6ecb486fa21eb007bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 17:41:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/60/670360/4/check/openstack-tox-cover/3656be3/cover/ : SUCCESS in 2m 51s\n- openstack-tox-lower-constraints https://logs.opendev.org/60/670360/4/check/openstack-tox-lower-constraints/abda31d/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://logs.opendev.org/60/670360/4/check/openstack-tox-pep8/84e5c5d/ : SUCCESS in 3m 00s\n- openstack-tox-py27 https://logs.opendev.org/60/670360/4/check/openstack-tox-py27/c793b12/ : SUCCESS in 2m 47s\n- openstack-tox-py36 https://logs.opendev.org/60/670360/4/check/openstack-tox-py36/664b195/ : SUCCESS in 3m 16s\n- openstack-tox-py37 https://logs.opendev.org/60/670360/4/check/openstack-tox-py37/536950a/ : SUCCESS in 3m 53s\n- openstack-tox-docs https://logs.opendev.org/60/670360/4/check/openstack-tox-docs/bf9000d/html/ : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://logs.opendev.org/60/670360/4/check/build-openstack-releasenotes/2e7361a/html/ : SUCCESS in 3m 06s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/60/670360/4/check/sushy-tempest-ironic-partition-redfish-src/77a85f1/ : SUCCESS in 41m 29s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/60/670360/4/check/sushy-tempest-ironic-partition-redfish-src-python2/222fb43/ : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":4},{"id":"5e7d8a9b357897b7ab78acb16b2689a3d24f19f6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-13 20:17:41.000000000","message":"Patch Set 4: Code-Review+1\n\nQuick pass, basically lgtm, but need to wrap my head around it.","accounts_in_message":[],"_revision_number":4},{"id":"f4ce176510667ca5d70f96fd47e06e354b5bfce3","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-14 10:21:20.000000000","message":"Patch Set 4:\n\n\u003e Do you feel I do need to split it out or is it OK to leave in this patch?\n\nWell, in my opinion it makes git history easier to work with if we\u0027d have those two features separate. But I am OK with whatever choice you take.","accounts_in_message":[],"_revision_number":4},{"id":"6298797576566cae4cb9bccf4b01a7b3ac300823","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-14 10:24:32.000000000","message":"Patch Set 4:\n\nMay be we could put example use into a docstring?\n\nBTW, how about adding a doc hook into `doc/source/reference/api`?","accounts_in_message":[],"_revision_number":4},{"id":"9528be6ac6897840cdc47573a4d4c3e1b5f92b44","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-21 17:35:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0f37f1df5712c47951835c9a24a46d60603a42d7","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-22 05:35:04.000000000","message":"Patch Set 4:\n\nIf we could wait merging this till next week, I plan on using this code for a real thing.","accounts_in_message":[],"_revision_number":4},{"id":"cb576271aa30c725b87f5a1315e6546b38758acb","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-22 13:05:30.000000000","message":"Patch Set 4: Code-Review+1\n\nOkay Ilya, ping me when another review is needed. Lowering to +1.","accounts_in_message":[],"_revision_number":4},{"id":"b1f8935f3898d7d714d3b9ebe83bcc4b8d50db7b","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 13:56:31.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e9899e04f0d6bc4d2069229211f847c3b4488bdf","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 14:27:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0651355e5d2dc191804bc71467eba11f5e477df5","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 15:33:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e48e47cd7b3c68c90ca42921f4a88848671d452a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 16:38:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9e34dcc8bc923c40488e6ceac3dc03a6ad452307","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 17:10:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3af723f6378f57802942889f444ec4799658427","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 17:41:02.000000000","message":"Patch Set 4:\n\nMy simple and primitive take [1][2] on the task monitor. I hope this explains my use-case.\n\n1. https://github.com/etingof/sushy-oem-dellemc/blob/master/sushy_oem_dellemc/resources/manager/manager.py#L95\n2. https://github.com/etingof/sushy-oem-dellemc/blob/master/sushy_oem_dellemc/asynchronous.py#L33","accounts_in_message":[],"_revision_number":4},{"id":"e5ec6ab6e5ea5d553dc76f51214569fa82730552","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-09-10 16:17:55.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nIlya: Thanks for the detailed comments. I\u0027ve added some responses and will start working on updates for a new patch set...","accounts_in_message":[],"_revision_number":4},{"id":"5df7c1e53fa6413a49445ca43f31caf77894b40c","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-09-11 05:21:39.000000000","message":"Patch Set 4:\n\nHi Bill,\n\nJust a quick update on the use-case [1] I work on. The whole thing is kind of operational by now, it may better illustrate possibly API needs for the TaskMonitor.\n\n1. https://github.com/etingof/sushy-oem-dellemc/blob/master/sushy_oem_dellemc/asynchronous.py","accounts_in_message":[],"_revision_number":4},{"id":"1eb314f5c6f488ddc47d3019392ff50246ba29a0","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-09-12 17:18:06.000000000","message":"Patch Set 4:\n\nHi, Ilya. Thanks for that. I will use that use-case for reference while making the next patch set updates.","accounts_in_message":[],"_revision_number":4},{"id":"7ce3d8aba000538cf9cbe5f2565da3aed655d1eb","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-10-14 22:41:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7ffd8d2a7d4e11ca85c99bfef87adafefaf29db4","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-10-14 22:46:10.000000000","message":"Patch Set 4:\n\n(5 comments)\n\nIlya - I hope the updates in patch set 5 are close to what you described in your comments and the use case you pointed me to.\n\nI also reverted the changes to volume per your suggestion, so this is now just a patch for the TaskMonitor.","accounts_in_message":[],"_revision_number":4},{"id":"49d5fc952f1c8f6ec24f3289ef89ee700f958820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 23:26:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fd30c87c9504e7ead16b1d917874fd1 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7ab8ab3dacf4c9fba1fc551fc0ab3aa : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae2e97097944608a7c8106f50bfb8ea : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f36b9e4ac68947e5917017bc0fa7f77a : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6c5c5f6cb8471cb0cc40ce1ec4828d : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bebc7d3e14c644ba98a54a13dd65f74d : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c395fbc460b41fba1f043ba3b0d3141 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b59c0469f94f41ab4a46794b4bc0ad : SUCCESS in 3m 17s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/5da86a6b745447bcb6a29bc805ddf2cb : SUCCESS in 41m 41s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/361ac8c2380043839f638f3429871b5a : SUCCESS in 43m 32s","accounts_in_message":[],"_revision_number":5},{"id":"a00e976ade814d83a155f36963c2c22b830e4f3f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-31 18:04:32.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d805e3a8c944d6da22396eceef83ca5f12dae253","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 11:44:59.000000000","message":"Patch Set 5:\n\n(9 comments)\n\nLooks good! A few nits inline.","accounts_in_message":[],"_revision_number":5},{"id":"405218396e4203afdf731bbe00e2fc80b2efafba","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-11-05 21:18:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"720e58fcb8d07a8c65de5acbb28137d58ddc16f2","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2019-11-05 21:26:24.000000000","message":"Patch Set 6:\n\n(9 comments)\n\nIlya and Julia: thanks for the reviews!\n\nIlya: I added 2 instances of \u0027raise ConnectionError()\u0027 in connector.py (for the cases of a 202 status but no Location header and for a timeout in the blocking case). The ConnectionError seemed the best fit to me. But if you have a better idea for the exception let me know.","accounts_in_message":[],"_revision_number":6},{"id":"79dd8f6d13a086f3e4e859bb032cc97c5d311b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 22:20:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b4c7d94c93b472aa3b469ee88609768 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18802dfa038b4791a8193031d0225f93 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a78268caedd441895b22c19f8f6a1c3 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46ab618101124bfa9cf2327f358090b6 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/968178e1427c4d018bdd0989d6699290 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39c1aa3be8104035817bd8c5b471a0ce : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f6046041744db881dd27fb398c0686 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fd572685f7c4cef9168bcbd3fbd48fa : SUCCESS in 3m 14s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9aeb216624af4cc4970c33c4ef2f3664 : SUCCESS in 56m 53s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/f0d485ac9230464688a87cc436afe79c : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":6},{"id":"ad6dacc708e9f630d6911e6cd2738c5a0b7f2710","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-15 19:39:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"998bf876c29b4612eecba8eef1360480fde787f6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-15 20:23:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-15:13:39/670360/7/ : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":7},{"id":"ef610e07ec23e4a5e1cb956b49af351b1f62cccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 20:24:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb23bb1483fd4beb9e7686bc17414769 : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d1a053473942e686c061e754758ec0 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de46763fc2984225a721d8e8ee229e42 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/544446d7b78b475c86615100b8fee77a : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0dd57d2c0794cb5ba017197ca069af5 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c515f8d9c2c4184a3dc14abba23bf1b : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b5f223489e54e4d83e28763275f91c2 : SUCCESS in 5m 52s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/46b4ea25b8d84e3fa7d3aac545bfb3b1 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":7},{"id":"ef1ce5ea6140d84c3fef063dad017eeed4293512","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-01-17 21:16:01.000000000","message":"Patch Set 7:\n\ndell-recheck","accounts_in_message":[],"_revision_number":7},{"id":"f2e904e825abe73bbb47a5c3295efdebde4d61fa","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-17 21:17:57.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-17:15:16/670360/7/ : FAILURE in 1m 07s","accounts_in_message":[],"_revision_number":7},{"id":"d0cdcb7eb96fe96f0cd3fd290f4368e97f5fd3c2","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-17 21:25:16.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-17:15:24/670360/7/ : FAILURE in 5s","accounts_in_message":[],"_revision_number":7},{"id":"f92c37326cd694547f479d0559a41ee3d75c0c28","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-17 21:26:04.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-17:15:25/670360/7/ : FAILURE in 6s","accounts_in_message":[],"_revision_number":7},{"id":"d956237013fdbffee9932568f1a43bac8402bac5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-17 21:28:07.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-17:15:27/670360/7/ : FAILURE in 6s","accounts_in_message":[],"_revision_number":7},{"id":"cbf14cba2b5f5337dcb51ea4da6c1ec36e8d8c7b","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-01-20 21:28:43.000000000","message":"Patch Set 7:\n\ndell-recheck","accounts_in_message":[],"_revision_number":7},{"id":"8c5211d8fce8ec125bae05a2a91d5de330056f54","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-21 01:35:53.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-20:15:28/670360/7/ : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":7},{"id":"a78eaeabbe54416b1a0b74817c08fa2f24d94bf6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-27 13:56:13.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":7},{"id":"4e044f0acdc3f015a1844b1db4208059c413dd97","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-27 14:23:57.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"10a418ef882085cb3249f4890e0c6f673bff5bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:24:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"335e619e611e4ff915c211bfee31f6ad84fda83e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:25:50.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a4f630810f8400c8bd9380cfb27b996 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97cc23e875ec414fb79ea5dc822fc82b : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a158f2c9e37f4cccb97f2ff6abcb2d6c : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72187c4e563d47f9bb28eb24e45ee663 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d88169157f4aa4949fddef216f12ef : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/275c024fffca405689f319ef4f6a8662 : SUCCESS in 5m 24s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/47d5dacb38dc463a90494efa0a6e1021 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":7},{"id":"89c2a2d3fa653111bd5f9df8f159cccb75af21f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:25:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f0d1012a2ce2fb07cd6fbe3e20aeed45b874732c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:28:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f64328f505422d9d98380a2f68d4e6 : SUCCESS in 2m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78f8f5ffa7b544a6a794d02f2400b90b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1f60044bd860ff7efe47851b92eea695a755e28f","revisions":{"b9f71bfbcb2a9b9bc250e2effb259df58c57f851":{"kind":"REWORK","_number":1,"created":"2019-07-11 18:21:12.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/60/670360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/1"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Volume and VolumeCollection classes are also updated to take\nadvantage of the Task Monitor when performing the initialize_volume()\nand create_volume() methods respectively. Other sushy resource classes\ncan be similarly updated as needed in future updates.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b9f71bfbcb2a9b9bc250e2effb259df58c57f851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b9f71bfbcb2a9b9bc250e2effb259df58c57f851"}]},"branch":"refs/heads/master"},"2e5bd26ed68cc703b810c4d12e5327a7be002d97":{"kind":"REWORK","_number":2,"created":"2019-07-11 19:09:08.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/60/670360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/2"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 19:07:27.000000000","tz":-300},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Volume and VolumeCollection classes are also updated to take\nadvantage of the Task Monitor when performing the initialize_volume()\nand create_volume() methods respectively. Other sushy resource classes\ncan be similarly updated as needed in future updates.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2e5bd26ed68cc703b810c4d12e5327a7be002d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2e5bd26ed68cc703b810c4d12e5327a7be002d97"}]},"branch":"refs/heads/master"},"3379ae1d58c33c789485f2aba3236524c6d8b314":{"kind":"REWORK","_number":3,"created":"2019-07-18 23:46:11.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/60/670360/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/3"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-18 23:41:31.000000000","tz":-300},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Volume and VolumeCollection classes are also updated to take\nadvantage of the Task Monitor when performing the initialize_volume()\nand create_volume() methods respectively. Other sushy resource classes\ncan be similarly updated as needed in future updates.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3379ae1d58c33c789485f2aba3236524c6d8b314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3379ae1d58c33c789485f2aba3236524c6d8b314"}]},"branch":"refs/heads/master"},"d93114fa07674f440263ac63dccfc1430337a6fe":{"kind":"REWORK","_number":4,"created":"2019-08-08 16:42:40.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/60/670360/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/4"}}},"commit":{"parents":[{"commit":"dc127c80e514f60e0567f82a1d31c4d66c5e35e9","subject":"Refactor DurableName identifier and Protocol fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc127c80e514f60e0567f82a1d31c4d66c5e35e9"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-08-08 16:39:26.000000000","tz":-300},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Volume and VolumeCollection classes are also updated to take\nadvantage of the Task Monitor when performing the initialize_volume()\nand create_volume() methods respectively. Other sushy resource classes\ncan be similarly updated as needed in future updates.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d93114fa07674f440263ac63dccfc1430337a6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d93114fa07674f440263ac63dccfc1430337a6fe"}]},"branch":"refs/heads/master"},"1f91497a3d091e78735e6194720851215c736575":{"kind":"REWORK","_number":5,"created":"2019-10-14 22:41:26.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/60/670360/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/5"}}},"commit":{"parents":[{"commit":"6118000b749acff6690562bed20c5ee006b96650","subject":"Add versions to release notes series","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6118000b749acff6690562bed20c5ee006b96650"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-10-14 22:35:45.000000000","tz":-300},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Connector class is updated allow an option to make HTTP requests\nblocking. This change utilizes the newly added TaskMonitor.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1f91497a3d091e78735e6194720851215c736575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1f91497a3d091e78735e6194720851215c736575"}]},"branch":"refs/heads/master"},"6d3f52d32ba0ab3228caccde7639f68a2caa7479":{"kind":"REWORK","_number":6,"created":"2019-11-05 21:18:43.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/60/670360/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/6"}}},"commit":{"parents":[{"commit":"5990b5df78808bb53bc5b64c9c4160f8748b055d","subject":"Switch jobs to python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5990b5df78808bb53bc5b64c9c4160f8748b055d"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-11-05 21:15:18.000000000","tz":-360},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Connector class is updated allow an option to make HTTP requests\nblocking. This change utilizes the newly added TaskMonitor.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6d3f52d32ba0ab3228caccde7639f68a2caa7479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6d3f52d32ba0ab3228caccde7639f68a2caa7479"}]},"branch":"refs/heads/master"},"1f60044bd860ff7efe47851b92eea695a755e28f":{"kind":"REWORK","_number":7,"created":"2020-01-15 19:39:44.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/60/670360/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/60/670360/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/60/670360/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/60/670360/7"}}},"commit":{"parents":[{"commit":"c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2019-07-11 18:17:08.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-01-15 19:39:44.000000000","tz":-360},"subject":"Add Task Monitor support for async operations","message":"Add Task Monitor support for async operations\n\nRedfish services that support asynchronous operations implement the\nTask service, Task resource, and Task Monitor. When performing long\nrunning update operations, the service returns a status of 202\nAccepted. When returning a status code of 202, the response includes\na Location header containing the URL of the Task Monitor. The response\nmay also include the Retry-After header to specify the amount of time\nthe client should wait before querying the operation status.\n\nThe Task Monitor is an opaque URL that the client can query via a GET\nrequest. While the async operation is still in progress, the service\nreturns a 202 Accepted status. The body of the response may be a Task\nresource or it may be empty.\n\nOnce the operation completes, a GET on the Task Monitor will return a\nstatus that indicates the result of the operation (e.g., 200, 201,\n4XX).\n\nA new TaskMonitor class is added to represent the Task Monitor and\ngive the sushy user the ability to monitor the status of an async\noperation.\n\nThe Connector class is updated allow an option to make HTTP requests\nblocking. This change utilizes the newly added TaskMonitor.\n\nChange-Id: Ice6700806710f37b590f6b879ec38656326b39c6\nStory: 2003514\nTask: 30719\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1f60044bd860ff7efe47851b92eea695a755e28f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1f60044bd860ff7efe47851b92eea695a755e28f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
