)]}'
{"id":"openstack%2Fsushy~671306","triplet_id":"openstack%2Fsushy~master~Ie9c2f37062b8a8c9a1a082217dc22c79c5327fba","project":"openstack/sushy","branch":"master","topic":"03-add-vmedia-eject-param","hashtags":[],"change_id":"Ie9c2f37062b8a8c9a1a082217dc22c79c5327fba","subject":"Retry Virtual Media eject action on HTTP 400 response","status":"MERGED","created":"2019-07-17 14:17:59.000000000","updated":"2019-08-01 08:46:25.000000000","submitted":"2019-08-01 08:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671306-1564649185452-2c0f3d06","meta_rev_id":"23d215b4e3749f5854a607e75adcdd0b93ca45c5","_number":671306,"virtual_id_number":671306,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-08-01 08:46:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":2,"date":"2019-07-31 13:07:36.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2019-07-31 13:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-01 07:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-01 07:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 07:14:38.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-07-19 08:19:16.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-08-01 07:07:02.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-08-01 08:46:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be430f554ac2630006d3749b48219946c2848b62","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-17 14:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3903891384699526a9e0b2c3740a6af532e35bfe","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-17 14:19:22.000000000","message":"Patch Set 1:\n\nShould we always enclose an empty JSON document with POST? Why not?","accounts_in_message":[],"_revision_number":1},{"id":"00c84584c55b33991fe0befe1e84b4f2388aa201","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:14:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/06/671306/1/check/openstack-tox-cover/dd9c9cf/cover/ : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/671306/1/check/openstack-tox-lower-constraints/f48b711/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/06/671306/1/check/openstack-tox-pep8/5d51c3e/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/06/671306/1/check/openstack-tox-py27/925fc4e/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/06/671306/1/check/openstack-tox-py36/67539be/ : SUCCESS in 2m 50s\n- openstack-tox-py37 http://logs.openstack.org/06/671306/1/check/openstack-tox-py37/dd1edfa/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/06/671306/1/check/openstack-tox-docs/2392e5e/html/ : SUCCESS in 3m 16s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/06/671306/1/check/sushy-tempest-ironic-partition-redfish-src/1dfe346/ : SUCCESS in 38m 28s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/06/671306/1/check/sushy-tempest-ironic-partition-redfish-src-python2/d979efd/ : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ee983c06976129c73340d03041ae0d317fab3174","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-07-18 07:14:38.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you Ilya.","accounts_in_message":[],"_revision_number":1},{"id":"997a27fae957929034222121eda45e37f26ca079","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-19 08:19:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5240f2ffab6300b4a33da08807608958fea997ab","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-07-22 05:38:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fae736c84d11c6e52168455aee0e92d28fa8fa25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 06:23:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/06/671306/1/check/openstack-tox-cover/60ac4b4/cover/ : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/671306/1/check/openstack-tox-lower-constraints/ed91108/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/06/671306/1/check/openstack-tox-pep8/bea0ace/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/06/671306/1/check/openstack-tox-py27/84610d2/ : SUCCESS in 2m 35s\n- openstack-tox-py36 http://logs.openstack.org/06/671306/1/check/openstack-tox-py36/54307cb/ : SUCCESS in 7m 12s\n- openstack-tox-py37 http://logs.openstack.org/06/671306/1/check/openstack-tox-py37/2c815f6/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/06/671306/1/check/openstack-tox-docs/530dcb6/html/ : SUCCESS in 6m 32s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/06/671306/1/check/sushy-tempest-ironic-partition-redfish-src/a93ada1/ : SUCCESS in 42m 47s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/06/671306/1/check/sushy-tempest-ironic-partition-redfish-src-python2/b840a5a/ : FAILURE in 28m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e7d95b346bd0320a3642877f0d4a20537b3b87d8","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-22 07:58:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e13b0ca1659002edcbe07867b83d89f3959901e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 08:49:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/06/671306/2/check/openstack-tox-cover/82abfaa/cover/ : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/671306/2/check/openstack-tox-lower-constraints/f1bbd15/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/06/671306/2/check/openstack-tox-pep8/4de8d34/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/06/671306/2/check/openstack-tox-py27/2c6ddeb/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/06/671306/2/check/openstack-tox-py36/16a7cef/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/06/671306/2/check/openstack-tox-py37/775f84b/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/06/671306/2/check/openstack-tox-docs/5d311bc/html/ : SUCCESS in 3m 19s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/06/671306/2/check/sushy-tempest-ironic-partition-redfish-src/2233086/ : SUCCESS in 42m 44s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/06/671306/2/check/sushy-tempest-ironic-partition-redfish-src-python2/6963415/ : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c186574c03330c8d57266b085172bae94c91d6f7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-07-23 09:27:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1831d9f94c879bd8a12efae78652e23a2583030f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-07-31 13:07:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d56623d5125e5de756dc1a27e4bdcd0a6f93a5bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 14:05:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/06/671306/3/check/openstack-tox-cover/b0a1466/cover/ : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/06/671306/3/check/openstack-tox-lower-constraints/8f5a67d/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/06/671306/3/check/openstack-tox-pep8/27abab8/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/06/671306/3/check/openstack-tox-py27/2459d97/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/06/671306/3/check/openstack-tox-py36/5190ef6/ : SUCCESS in 2m 44s\n- openstack-tox-py37 http://logs.openstack.org/06/671306/3/check/openstack-tox-py37/0224dd9/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/06/671306/3/check/openstack-tox-docs/c6a71e5/html/ : SUCCESS in 4m 08s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/06/671306/3/check/sushy-tempest-ironic-partition-redfish-src/9950e91/ : SUCCESS in 46m 10s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/06/671306/3/check/sushy-tempest-ironic-partition-redfish-src-python2/f0ccbc3/ : FAILURE in 55m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d91281aeceba9811928500c2cafc2d24e262a295","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-01 07:07:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"563228ddaa14eb7cc1a3d79479a686d483e0fb7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 07:58:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/06/671306/3/check/openstack-tox-cover/75455b8/cover/ : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/06/671306/3/check/openstack-tox-lower-constraints/e559408/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://logs.opendev.org/06/671306/3/check/openstack-tox-pep8/cd16598/ : SUCCESS in 3m 17s\n- openstack-tox-py27 https://logs.opendev.org/06/671306/3/check/openstack-tox-py27/6c91d98/ : SUCCESS in 3m 05s\n- openstack-tox-py36 https://logs.opendev.org/06/671306/3/check/openstack-tox-py36/fea57a2/ : SUCCESS in 3m 12s\n- openstack-tox-py37 https://logs.opendev.org/06/671306/3/check/openstack-tox-py37/7be62d7/ : SUCCESS in 3m 37s\n- openstack-tox-docs https://logs.opendev.org/06/671306/3/check/openstack-tox-docs/ee0aa84/html/ : SUCCESS in 3m 06s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/06/671306/3/check/sushy-tempest-ironic-partition-redfish-src/61fba99/ : SUCCESS in 45m 49s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/06/671306/3/check/sushy-tempest-ironic-partition-redfish-src-python2/bf5b8f0/ : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":3},{"id":"077146bb81d3526084b3e9d8f42bcab503d0fac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 07:59:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"589791513be3167bc3fcd122f88b73f4e827e781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 08:46:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"58e4a889a8f90e0fd172e8c7326216f2d4e75b6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 08:46:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/06/671306/3/gate/openstack-tox-lower-constraints/12699b1/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://logs.opendev.org/06/671306/3/gate/openstack-tox-pep8/0a489f5/ : SUCCESS in 3m 38s\n- openstack-tox-py27 https://logs.opendev.org/06/671306/3/gate/openstack-tox-py27/bf88344/ : SUCCESS in 3m 53s\n- openstack-tox-py36 https://logs.opendev.org/06/671306/3/gate/openstack-tox-py36/92363c1/ : SUCCESS in 2m 58s\n- openstack-tox-py37 https://logs.opendev.org/06/671306/3/gate/openstack-tox-py37/dc551d8/ : SUCCESS in 3m 45s\n- openstack-tox-docs https://logs.opendev.org/06/671306/3/gate/openstack-tox-docs/59c46fb/html/ : SUCCESS in 3m 01s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/06/671306/3/gate/sushy-tempest-ironic-partition-redfish-src/345b2d9/ : SUCCESS in 41m 07s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/06/671306/3/gate/sushy-tempest-ironic-partition-redfish-src-python2/f754467/ : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"219d93aecd534b7a75dba49507523127dbde45b6","revisions":{"f7cef67b7a4cb166e3f5f8aed6ae2c369689c325":{"kind":"REWORK","_number":1,"created":"2019-07-17 14:17:59.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/06/671306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/671306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/671306/1"}}},"commit":{"parents":[{"commit":"b185444e4af93ea8096f6b76a59d25f1ed6651f8","subject":"Cache message registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b185444e4af93ea8096f6b76a59d25f1ed6651f8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-17 14:17:07.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-17 14:17:07.000000000","tz":120},"subject":"Retry Virtual Media eject action on HTTP 400 response","message":"Retry Virtual Media eject action on HTTP 400 response\n\nWhen attempting to eject virtual media and HTTP server responds\nwith HTTP error 400, retry the same HTTP query with empty\nJSON document.\n\nThis change simply extends HTTP error code list when this\nretry should happen to account for Dell server(s) - unlike\nHPE ones (which return error 415), Dell returns 400.\n\nSee also: https://review.opendev.org/#/c/636845/\n\nChange-Id: Ie9c2f37062b8a8c9a1a082217dc22c79c5327fba\nStory: 2004995\nTask: 35845\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f7cef67b7a4cb166e3f5f8aed6ae2c369689c325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f7cef67b7a4cb166e3f5f8aed6ae2c369689c325"}]},"branch":"refs/heads/master"},"680d4e44830718883ac2579806a559b26425fb26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-22 07:58:44.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/06/671306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/671306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/671306/2"}}},"commit":{"parents":[{"commit":"a009312bb71b925699f0474ee6f808ba9f993f65","subject":"Cache message registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a009312bb71b925699f0474ee6f808ba9f993f65"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-17 14:17:07.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-22 07:58:36.000000000","tz":120},"subject":"Retry Virtual Media eject action on HTTP 400 response","message":"Retry Virtual Media eject action on HTTP 400 response\n\nWhen attempting to eject virtual media and HTTP server responds\nwith HTTP error 400, retry the same HTTP query with empty\nJSON document.\n\nThis change simply extends HTTP error code list when this\nretry should happen to account for Dell server(s) - unlike\nHPE ones (which return error 415), Dell returns 400.\n\nSee also: https://review.opendev.org/#/c/636845/\n\nChange-Id: Ie9c2f37062b8a8c9a1a082217dc22c79c5327fba\nStory: 2004995\nTask: 35845\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/680d4e44830718883ac2579806a559b26425fb26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/680d4e44830718883ac2579806a559b26425fb26"}]},"branch":"refs/heads/master"},"219d93aecd534b7a75dba49507523127dbde45b6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-31 13:07:36.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/06/671306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/671306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/671306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/671306/3"}}},"commit":{"parents":[{"commit":"ba1b8220e7d2de2f689ab014284e938ee8190709","subject":"Cache message registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ba1b8220e7d2de2f689ab014284e938ee8190709"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-17 14:17:07.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-07-31 13:07:24.000000000","tz":120},"subject":"Retry Virtual Media eject action on HTTP 400 response","message":"Retry Virtual Media eject action on HTTP 400 response\n\nWhen attempting to eject virtual media and HTTP server responds\nwith HTTP error 400, retry the same HTTP query with empty\nJSON document.\n\nThis change simply extends HTTP error code list when this\nretry should happen to account for Dell server(s) - unlike\nHPE ones (which return error 415), Dell returns 400.\n\nSee also: https://review.opendev.org/#/c/636845/\n\nChange-Id: Ie9c2f37062b8a8c9a1a082217dc22c79c5327fba\nStory: 2004995\nTask: 35845\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/219d93aecd534b7a75dba49507523127dbde45b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/219d93aecd534b7a75dba49507523127dbde45b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
