)]}'
{"id":"openstack%2Fsushy~671413","triplet_id":"openstack%2Fsushy~master~I8d234b3e562a46d3be27f6eae83b6f2e99813c83","project":"openstack/sushy","branch":"master","topic":"story-2006246","hashtags":[],"change_id":"I8d234b3e562a46d3be27f6eae83b6f2e99813c83","subject":"Action #Bios.ResetBios fails as POST request has no body","status":"MERGED","created":"2019-07-18 07:05:48.000000000","updated":"2019-08-02 21:56:24.000000000","submitted":"2019-08-02 21:56:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671413-1564782984155-a2196f10","meta_rev_id":"fb9f8a13518cb8fd113aa9c2761ce80ae580d39e","_number":671413,"virtual_id_number":671413,"owner":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-08-02 21:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2019-08-02 21:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-22 08:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2019-08-02 21:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 08:14:10.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-08-02 21:07:53.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-02 21:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17866c3d5a73705c5295fc4ccb8ba106112f2741","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-07-18 07:05:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"646f1d58dcd2631a18c69616e57026400eb2ba19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 07:51:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/13/671413/1/check/openstack-tox-cover/d657296/cover/ : SUCCESS in 7m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/671413/1/check/openstack-tox-lower-constraints/f66410d/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/13/671413/1/check/openstack-tox-pep8/563e906/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/13/671413/1/check/openstack-tox-py27/06ea8bd/ : SUCCESS in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/13/671413/1/check/openstack-tox-py36/be96fdd/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/13/671413/1/check/openstack-tox-py37/ad223d0/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/13/671413/1/check/openstack-tox-docs/050d15b/html/ : SUCCESS in 3m 31s\n- build-openstack-releasenotes http://logs.openstack.org/13/671413/1/check/build-openstack-releasenotes/ce111d4/html/ : SUCCESS in 4m 00s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/13/671413/1/check/sushy-tempest-ironic-partition-redfish-src/bdbb77b/ : SUCCESS in 39m 49s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/13/671413/1/check/sushy-tempest-ironic-partition-redfish-src-python2/6e4681b/ : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4ac719d6182bb748acf6306397882e3559e8ed7e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-21 14:37:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8ce1b1de3e8778a5f2e0fcd4b6d786e8966c1587","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-07-22 04:37:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e068b46b6df7b3f98c79d8e955f5bc41c3f86e36","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-07-22 04:38:28.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThank you Dmitry.","accounts_in_message":[],"_revision_number":1},{"id":"a66283fefcc5d68465c0cd3c5902e983f5145ef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 06:22:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/13/671413/2/check/openstack-tox-cover/7536092/cover/ : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/671413/2/check/openstack-tox-lower-constraints/e615ed1/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/13/671413/2/check/openstack-tox-pep8/68c53e7/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/13/671413/2/check/openstack-tox-py27/c73bbc5/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/13/671413/2/check/openstack-tox-py36/c0fb41e/ : SUCCESS in 3m 29s\n- openstack-tox-py37 http://logs.openstack.org/13/671413/2/check/openstack-tox-py37/0c7428b/ : SUCCESS in 3m 19s\n- openstack-tox-docs http://logs.openstack.org/13/671413/2/check/openstack-tox-docs/154a8ab/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/13/671413/2/check/build-openstack-releasenotes/98ba8d9/html/ : SUCCESS in 3m 42s\n- sushy-tempest-ironic-partition-redfish-src http://logs.openstack.org/13/671413/2/check/sushy-tempest-ironic-partition-redfish-src/d303dfc/ : SUCCESS in 49m 12s\n- sushy-tempest-ironic-partition-redfish-src-python2 http://logs.openstack.org/13/671413/2/check/sushy-tempest-ironic-partition-redfish-src-python2/f6344c3/ : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e7bcd050a59c1faf8d79f02b2379c6f1fa350041","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-22 08:14:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"47f58fa4da9963c5dc6cb3da39ece239f5bf8608","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-02 21:07:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01809b1637fb24039927e92807d76b9b53d6f73e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 21:08:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c4c61d75d8b1164214cf67d7e8a98d617437d6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 21:56:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/13/671413/2/gate/openstack-tox-lower-constraints/a55f587/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://logs.opendev.org/13/671413/2/gate/openstack-tox-pep8/d4e399b/ : SUCCESS in 2m 48s\n- openstack-tox-py27 https://logs.opendev.org/13/671413/2/gate/openstack-tox-py27/012ade2/ : SUCCESS in 2m 41s\n- openstack-tox-py36 https://logs.opendev.org/13/671413/2/gate/openstack-tox-py36/d89dbf3/ : SUCCESS in 2m 58s\n- openstack-tox-py37 https://logs.opendev.org/13/671413/2/gate/openstack-tox-py37/b76da45/ : SUCCESS in 3m 04s\n- openstack-tox-docs https://logs.opendev.org/13/671413/2/gate/openstack-tox-docs/00ae8b2/html/ : SUCCESS in 2m 19s\n- build-openstack-releasenotes https://logs.opendev.org/13/671413/2/gate/build-openstack-releasenotes/9c20ec2/html/ : SUCCESS in 3m 06s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/13/671413/2/gate/sushy-tempest-ironic-partition-redfish-src/b36cb0f/ : SUCCESS in 38m 33s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/13/671413/2/gate/sushy-tempest-ironic-partition-redfish-src-python2/ca2fc50/ : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":2},{"id":"722cd649370ae14a9b66b3fdac18773ab89819d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 21:56:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83a6a8cf9173f612f66c34cbf20029d5ef216158","revisions":{"2e94607e92b47a93e44954ed108727b62843c642":{"kind":"REWORK","_number":1,"created":"2019-07-18 07:05:48.000000000","uploader":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"ref":"refs/changes/13/671413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/13/671413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/671413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/671413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/671413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/13/671413/1"}}},"commit":{"parents":[{"commit":"83c1a332f2c4982499a1b5b25657f8200edbdbed","subject":"Merge \"Make UpdateService.simple_update() operational\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/83c1a332f2c4982499a1b5b25657f8200edbdbed"}]}],"author":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2019-07-18 08:32:38.000000000","tz":-240},"committer":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2019-07-18 08:32:38.000000000","tz":-240},"subject":"Action #Bios.ResetBios fails as POST request has no body","message":"Action #Bios.ResetBios fails as POST request has no body\n\nBios action #Bios.ResetBios fails with HTTP code 415 as BMC expects\nthe POST request with empty body instead of no body.\n\nChange-Id: I8d234b3e562a46d3be27f6eae83b6f2e99813c83\nStory: 2006246\nTask: 35854\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2e94607e92b47a93e44954ed108727b62843c642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2e94607e92b47a93e44954ed108727b62843c642"}]},"branch":"refs/heads/master"},"83a6a8cf9173f612f66c34cbf20029d5ef216158":{"kind":"REWORK","_number":2,"created":"2019-07-22 04:37:50.000000000","uploader":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"ref":"refs/changes/13/671413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/13/671413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/671413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/671413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/671413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/13/671413/2"}}},"commit":{"parents":[{"commit":"83c1a332f2c4982499a1b5b25657f8200edbdbed","subject":"Merge \"Make UpdateService.simple_update() operational\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/83c1a332f2c4982499a1b5b25657f8200edbdbed"}]}],"author":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2019-07-18 08:32:38.000000000","tz":-240},"committer":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2019-07-22 06:08:12.000000000","tz":-240},"subject":"Action #Bios.ResetBios fails as POST request has no body","message":"Action #Bios.ResetBios fails as POST request has no body\n\nBios action #Bios.ResetBios fails with HTTP code 415 as BMC expects\nthe POST request with empty body instead of no body.\n\nChange-Id: I8d234b3e562a46d3be27f6eae83b6f2e99813c83\nStory: 2006246\nTask: 35854\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/83a6a8cf9173f612f66c34cbf20029d5ef216158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/83a6a8cf9173f612f66c34cbf20029d5ef216158"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
