)]}'
{"id":"openstack%2Fsushy~674882","triplet_id":"openstack%2Fsushy~master~Idee0d17558e4332f44f623add69c00d4a4e2bdd4","project":"openstack/sushy","branch":"master","topic":"improve-oem-actions","hashtags":[],"change_id":"Idee0d17558e4332f44f623add69c00d4a4e2bdd4","subject":"Unify OEM Actions with non-OEM Actions","status":"MERGED","created":"2019-08-06 17:13:25.000000000","updated":"2019-08-12 15:07:39.000000000","submitted":"2019-08-12 15:07:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674882-1565622459494-ff059232","meta_rev_id":"747a7b67e714bf50abc4589dd17cb492be6c9eeb","_number":674882,"virtual_id_number":674882,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-08-12 15:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2019-08-12 14:11:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-07 08:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2019-08-12 14:11:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 08:31:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-08-12 14:11:06.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-08-12 15:07:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6e9295bb35a93e5e47cb02627e53bfc282af037","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-06 17:13:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82032ff72cfed64b6be8443c0b6ad3efbb26edd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:18:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/82/674882/1/check/openstack-tox-cover/c413843/cover/ : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/82/674882/1/check/openstack-tox-lower-constraints/ed52791/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://logs.opendev.org/82/674882/1/check/openstack-tox-pep8/742decd/ : SUCCESS in 3m 22s\n- openstack-tox-py27 https://logs.opendev.org/82/674882/1/check/openstack-tox-py27/b840089/ : SUCCESS in 3m 42s\n- openstack-tox-py36 https://logs.opendev.org/82/674882/1/check/openstack-tox-py36/565a457/ : SUCCESS in 3m 42s\n- openstack-tox-py37 https://logs.opendev.org/82/674882/1/check/openstack-tox-py37/8a20881/ : SUCCESS in 3m 38s\n- openstack-tox-docs https://logs.opendev.org/82/674882/1/check/openstack-tox-docs/cbb9031/html/ : SUCCESS in 2m 55s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/82/674882/1/check/sushy-tempest-ironic-partition-redfish-src/0316fac/ : SUCCESS in 46m 05s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/82/674882/1/check/sushy-tempest-ironic-partition-redfish-src-python2/e22f7d0/ : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":1},{"id":"804a561ccfe6b8e72582e16c92386938dbb1dd4e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-07 08:31:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"169d25522fc8614ed626212bde64397c5373bbb7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-12 14:11:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"488072831ff8dae8974ca6e34da7dd67f8a1697e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:11:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2587f562691d52b30a6502c9a83b6fa615396422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:07:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e1d0e8de8acb193eaf173ff1b3f74237b377006","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:07:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/82/674882/1/gate/openstack-tox-lower-constraints/dfc5134/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://logs.opendev.org/82/674882/1/gate/openstack-tox-pep8/e4b3ce7/ : SUCCESS in 2m 46s\n- openstack-tox-py27 https://logs.opendev.org/82/674882/1/gate/openstack-tox-py27/30e1b5d/ : SUCCESS in 3m 11s\n- openstack-tox-py36 https://logs.opendev.org/82/674882/1/gate/openstack-tox-py36/26a8889/ : SUCCESS in 3m 13s\n- openstack-tox-py37 https://logs.opendev.org/82/674882/1/gate/openstack-tox-py37/6049c0a/ : SUCCESS in 3m 32s\n- openstack-tox-docs https://logs.opendev.org/82/674882/1/gate/openstack-tox-docs/86ac73c/html/ : SUCCESS in 2m 53s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/82/674882/1/gate/sushy-tempest-ironic-partition-redfish-src/44de2e7/ : SUCCESS in 40m 30s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/82/674882/1/gate/sushy-tempest-ironic-partition-redfish-src-python2/58c71ac/ : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb9859da99ffd3f5b1f74d9ce43e606ea0e96153","revisions":{"bb9859da99ffd3f5b1f74d9ce43e606ea0e96153":{"kind":"REWORK","_number":1,"created":"2019-08-06 17:13:25.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/82/674882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/674882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/674882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/674882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/674882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/674882/1"}}},"commit":{"parents":[{"commit":"a18027797e07113d9b1dd4070669b3110bfc704c","subject":"Merge \"Retry Virtual Media eject action on HTTP 400 response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a18027797e07113d9b1dd4070669b3110bfc704c"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-06 17:12:48.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-06 17:12:48.000000000","tz":120},"subject":"Unify OEM Actions with non-OEM Actions","message":"Unify OEM Actions with non-OEM Actions\n\nThis change introduces a few OEM base types to align OEM Actions\nabstraction with non-OEM one. One of the immediate benefits is\nthat Action fields are actually parsed and assigned to sushy\nfields rather than represented as a dict.\n\nChange-Id: Idee0d17558e4332f44f623add69c00d4a4e2bdd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bb9859da99ffd3f5b1f74d9ce43e606ea0e96153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bb9859da99ffd3f5b1f74d9ce43e606ea0e96153"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
