)]}'
{"id":"openstack%2Fsushy~675073","triplet_id":"openstack%2Fsushy~master~I9da9a83b72b670e15b53ef906f186473340c0935","project":"openstack/sushy","branch":"master","topic":"02-add-key-matcher","hashtags":[],"change_id":"I9da9a83b72b670e15b53ef906f186473340c0935","subject":"Add conditional field matching","status":"MERGED","created":"2019-08-07 09:54:12.000000000","updated":"2019-09-04 13:45:38.000000000","submitted":"2019-09-04 13:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675073-1567604604418-c2cdff65","meta_rev_id":"8b63cf8ab3b431f93c7d63f6ccff64fd50e68883","_number":675073,"virtual_id_number":675073,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-09-04 13:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":2,"date":"2019-08-28 13:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2019-08-28 13:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":1,"date":"2019-09-04 12:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 15:20:56.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2019-08-22 08:33:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-09-04 12:52:41.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-09-04 13:43:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a10eae2847745f00d41d54ddb2c2ed0c61f626db","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-07 09:54:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36818ae38e02d2755172619d739a7b59fd057460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 10:47:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/73/675073/1/check/openstack-tox-cover/bc9c727/cover/ : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://logs.opendev.org/73/675073/1/check/openstack-tox-lower-constraints/58c3f77/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://logs.opendev.org/73/675073/1/check/openstack-tox-pep8/b8d7d0e/ : SUCCESS in 3m 46s\n- openstack-tox-py27 https://logs.opendev.org/73/675073/1/check/openstack-tox-py27/07f7094/ : SUCCESS in 3m 29s\n- openstack-tox-py36 https://logs.opendev.org/73/675073/1/check/openstack-tox-py36/78df242/ : SUCCESS in 4m 06s\n- openstack-tox-py37 https://logs.opendev.org/73/675073/1/check/openstack-tox-py37/4052dd6/ : SUCCESS in 4m 52s\n- openstack-tox-docs https://logs.opendev.org/73/675073/1/check/openstack-tox-docs/509c22d/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://logs.opendev.org/73/675073/1/check/build-openstack-releasenotes/0817a6a/html/ : SUCCESS in 4m 09s\n- sushy-tempest-ironic-partition-redfish-src https://logs.opendev.org/73/675073/1/check/sushy-tempest-ironic-partition-redfish-src/8918c2b/ : SUCCESS in 43m 50s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://logs.opendev.org/73/675073/1/check/sushy-tempest-ironic-partition-redfish-src-python2/65cae17/ : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"97a55daf0e1abeed10be4aba9d8e56f1918b6892","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-07 15:20:42.000000000","message":"Patch Set 1:\n\nThe use-case [1]\n\n1. https://github.com/etingof/sushy-oem-dellemc","accounts_in_message":[],"_revision_number":1},{"id":"0f89ffefcac403bfacd0a2a03464ccdc2ad99e2b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-13 20:33:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ea73197b06d7257b97f5b9164b6a7097899a01d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-16 12:50:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"807726bf068f67126995d4b2a56259538e3b50bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:44:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5476cd8471004461bdf4fb2b353063bc : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffe26c53349b46cda1b052b02e48e169 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c4878d448c400ab1930a619c239f35 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d7abc666d33410a8bb3326fc273371d : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0687cf769714cd5821b77f6b0a52743 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a4cd55743784d0090c2f7628a67d852 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3974f34db374143919c109aaa8ca59c : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1b31a06e05d431590330842cdd302f0 : SUCCESS in 4m 29s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/5c1ea2c45b744519bdef4ffab471b5da : SUCCESS in 51m 39s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/699a2e71edad42589ee0c8cc4573030c : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"f5c54c81114294770ffe2be99d514f9e3adf2b27","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-21 17:38:12.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"926a2ab7a225964846b63d53b405f80a49a16f7b","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-22 05:33:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e752f85226ee5fd2872da916b4ec5244fce95ae5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-08-22 08:33:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"315689612b65f87bb5b1496a407a6c7d38c46dae","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-26 10:42:14.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d1c40f5f4959d6535bbd89c18b8cfaac69691514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 11:40:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6f833eebf734e5aaab31ace776eabf2 : SUCCESS in 4m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d8973ecc4cf491e91bbbad6f8056122 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c65291d60e404fa6f6b9b7defac5b8 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da054a8959e14aab984fb55b0555d08c : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af548387a3a848e6b574c93409b7d440 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0e52a586e84492e9115c79cb96174aa : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce2589da4d5424fa84a01a0c99558d7 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2906c8d2efbe4d0aaacd6f8b3b38b43d : SUCCESS in 3m 55s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/c918536b7d974498a7b94ec3552e942e : SUCCESS in 41m 35s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/f4af638cc6a04c3995f8769442957559 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e1f910c1d913f9a1ea2474a65aac9e98850832da","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-27 18:16:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65dc853aa3384feea737de9dc52e77903e513227","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-08-27 18:16:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f84c43bf727b4f6f9752a62e5c7bf62f79bf62ac","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-08-28 13:45:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2e2ce0a8cdce40566830746ac5d5edfdff855067","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 14:38:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/626a47a432474145b66494b40330963c : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a294946ee43f4d6ab43fa04b352d0738 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1039a322d98c42e0957babff4d57f230 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9888c50041bb484d955f5d3ceafb4025 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cd71a161d8047d6834c9609a05748a7 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fda4a18ca5c340ff9efee534228e3429 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6f3065a5cd4bd28ddb1a6f54a6394f : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01f59d123b97400fbe304e78e705264a : SUCCESS in 4m 03s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0068c3c6ce13497d8ef48a95a8d9c601 : SUCCESS in 50m 14s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/0100ffcf7ca541bdb1090675f8d5b8cd : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":4},{"id":"c48a1994f680eee7763c4ab097a53322199cd6ef","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-09-04 12:52:41.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4789eac72ce27923433f68a0f42744f7efc02b1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:52:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0dd3aec7913e2bed275ff6d921aa4928af88a337","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:43:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f62285181c4ddfa51e4ff134c213d9 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7969de3bd1154a9aa0ec0e84c01af770 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b18364557854cc6b31caffb524457aa : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90f4fe644461453999885a5f89c8e0a7 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea549a85832d401697a3b097012198ae : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f8957c374b4cedad2df8c0b9774308 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03caa7dec9304b7b9692aaaa408c5c86 : SUCCESS in 4m 13s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/44eead20789949aa8e71f86c7e54366d : SUCCESS in 46m 23s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/050c766761a04982bcd9c888585c288e : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":4},{"id":"f5e196ded8c9f92edd04400f6740caca1974f4af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:43:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4e9c92a480e31aedac9301532b7917be3057a737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:45:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e6259782de47c4834cff2a44c5451e : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d6de035713a003d9bb74d7281b2026c8f5224be","revisions":{"bad28d6e09d3a08e4041a215be75b00b8ed779af":{"kind":"REWORK","_number":1,"created":"2019-08-07 09:54:12.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/73/675073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/73/675073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/73/675073/1"}}},"commit":{"parents":[{"commit":"bb9859da99ffd3f5b1f74d9ce43e606ea0e96153","subject":"Unify OEM Actions with non-OEM Actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bb9859da99ffd3f5b1f74d9ce43e606ea0e96153"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-07 09:51:59.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-07 09:53:12.000000000","tz":120},"subject":"Add conditional field matching","message":"Add conditional field matching\n\nAdds the ability to conditionally match sushy fields against\nreceived JSON object.\n\nThe conditional matching is performed by a user-supplied callable\nwhich gets the key to consider (along with the value and potentially\nother details) and should indicate the the caller if the match\noccurred.\n\nThe motivation behind this change is to accommodate malformed Redfish\nresource properties as observed in the OEM wilderness.\n\nChange-Id: I9da9a83b72b670e15b53ef906f186473340c0935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bad28d6e09d3a08e4041a215be75b00b8ed779af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bad28d6e09d3a08e4041a215be75b00b8ed779af"}]},"branch":"refs/heads/master"},"3267ecda2dfa0a774bf9272e74baf85ad85bedd0":{"kind":"REWORK","_number":2,"created":"2019-08-16 12:50:10.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/73/675073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/73/675073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/73/675073/2"}}},"commit":{"parents":[{"commit":"5b4219c0d4b3b3044bac8af816f1f6da3d2b415a","subject":"Change OEM extensions architecture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5b4219c0d4b3b3044bac8af816f1f6da3d2b415a"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-07 09:51:59.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-16 12:47:41.000000000","tz":120},"subject":"Add conditional field matching","message":"Add conditional field matching\n\nAdds the ability to conditionally match sushy fields against\nreceived JSON object.\n\nThe conditional matching is performed by a user-supplied callable\nwhich gets the key to consider (along with the value and potentially\nother details) and should indicate the the caller if the match\noccurred.\n\nThe motivation behind this change is to accommodate malformed Redfish\nresource properties as observed in the OEM wilderness.\n\nChange-Id: I9da9a83b72b670e15b53ef906f186473340c0935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3267ecda2dfa0a774bf9272e74baf85ad85bedd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3267ecda2dfa0a774bf9272e74baf85ad85bedd0"}]},"branch":"refs/heads/master"},"3cfa48026ce00d09e3d71c87ee67fac6ed89bf70":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-26 10:42:14.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/73/675073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/73/675073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/73/675073/3"}}},"commit":{"parents":[{"commit":"efaeaf9956b43e054c639929e0646161f22e84b1","subject":"Change OEM extensions architecture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/efaeaf9956b43e054c639929e0646161f22e84b1"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-07 09:51:59.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-26 10:42:06.000000000","tz":120},"subject":"Add conditional field matching","message":"Add conditional field matching\n\nAdds the ability to conditionally match sushy fields against\nreceived JSON object.\n\nThe conditional matching is performed by a user-supplied callable\nwhich gets the key to consider (along with the value and potentially\nother details) and should indicate the the caller if the match\noccurred.\n\nThe motivation behind this change is to accommodate malformed Redfish\nresource properties as observed in the OEM wilderness.\n\nChange-Id: I9da9a83b72b670e15b53ef906f186473340c0935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3cfa48026ce00d09e3d71c87ee67fac6ed89bf70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3cfa48026ce00d09e3d71c87ee67fac6ed89bf70"}]},"branch":"refs/heads/master"},"3d6de035713a003d9bb74d7281b2026c8f5224be":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-28 13:45:43.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/73/675073/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/73/675073/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/675073/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/73/675073/4"}}},"commit":{"parents":[{"commit":"e03caca38f72b778308d51e69856fc022dc32fd8","subject":"Change OEM extensions architecture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e03caca38f72b778308d51e69856fc022dc32fd8"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-07 09:51:59.000000000","tz":120},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2019-08-28 13:45:37.000000000","tz":120},"subject":"Add conditional field matching","message":"Add conditional field matching\n\nAdds the ability to conditionally match sushy fields against\nreceived JSON object.\n\nThe conditional matching is performed by a user-supplied callable\nwhich gets the key to consider (along with the value and potentially\nother details) and should indicate the the caller if the match\noccurred.\n\nThe motivation behind this change is to accommodate malformed Redfish\nresource properties as observed in the OEM wilderness.\n\nChange-Id: I9da9a83b72b670e15b53ef906f186473340c0935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d6de035713a003d9bb74d7281b2026c8f5224be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d6de035713a003d9bb74d7281b2026c8f5224be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
