)]}'
{"id":"openstack%2Fsushy~689835","triplet_id":"openstack%2Fsushy~master~Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6","project":"openstack/sushy","branch":"master","topic":"oem_extension_documentation","hashtags":[],"change_id":"Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6","subject":"Add OEM extension example script","status":"MERGED","created":"2019-10-21 16:18:48.000000000","updated":"2019-12-01 17:30:40.000000000","submitted":"2019-12-01 17:29:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":72,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689835-1575221367818-a4121484","meta_rev_id":"cb87d388e61fcb96da9b76e67fc91dfc922b720a","_number":689835,"virtual_id_number":689835,"owner":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-01 17:29:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"all":[{"value":2,"date":"2019-12-01 16:49:39.000000000","_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"all":[{"value":1,"date":"2019-12-01 16:49:44.000000000","_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 18:22:54.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2019-11-27 10:21:14.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-12-01 16:49:44.000000000","updated_by":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"reviewer":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"state":"REVIEWER"},{"updated":"2019-12-01 17:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6587eb5424f6adf798f4eb67b968c7e02aea4efd","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-21 16:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"702e7082899e558dcef083068fb506da389742c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 17:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c641650099034e62b505a9e739106758 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617c5accd6e34d70b549242dbeb285cb : FAILURE in 4m 23s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/6daa6560bc034be7ace1ef5ad577934b : SUCCESS in 58m 44s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/dfbfe764fb234409a193782874ab600d : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a164aed2522b4374d37a3499287110445fed60c4","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-21 20:19:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"32c74fdace70e25a352ab40906e6342aeea1b2c9","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-21 20:20:02.000000000","message":"Patch Set 1:\n\n...and `tox -e docs` as well.","accounts_in_message":[],"_revision_number":1},{"id":"0c903d6933beb15050c7237ddfb97e08eb2f6db4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-22 08:02:38.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nhey Khansa, this is a good start :)\nBe sure to check how report Storyboard stories in commit messages and remember to run tox locally to catch issues beforehand.","accounts_in_message":[],"_revision_number":1},{"id":"704c9352c93cb374eaaf71c0525fb822b43cb79a","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-23 15:37:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b7a89c081a53136bfb1d5bd44593d53bb949cca","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-23 15:46:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8980e07aa2cc95fbec30eda4db2bf5187689d4a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:30:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ac3313a25147f48e06d9361d956439 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d036342b2c4402e82a6ec35b04112b0 : SUCCESS in 6m 05s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/8e8a662bd1e24ef083e45efb7f605b95 : SUCCESS in 45m 23s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/f8b507aafe2f43ec84e87678c94fade8 : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9cf58b8a4f64010b4f49a87f7474a1aa3eba365a","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-23 16:34:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f337fe3a7d7f634752892693a04f2dc131c77cd6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-23 16:45:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"369f37ff60fc17e7e184f37d9ead8d8c1ddbeeb5","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-23 16:58:25.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5f1309fc1c206743ee7d8e2e426487c094a3891b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 17:50:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e97c58c38343f08f6583ea75515267 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b6520832db491080fdaff6f840d9b9 : SUCCESS in 5m 03s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/5491fa65dcde4c9594a29043e6473ee6 : SUCCESS in 43m 56s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/baccfe7e94304a759b44da9633109640 : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":4},{"id":"101e1d784e517a984c5a89515f7d5a4efdfa25bd","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-23 17:51:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c5604d74c0ed5ba00a9369a209258e85202ef309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:51:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8402a3a491d141faab53044c0627201b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/942594419b0b4faeb940713c0684dd74 : SUCCESS in 13m 51s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9d40658ae8d04c049cb802d5c61a59c6 : SUCCESS in 54m 03s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/801c5e72813e403db1c7f0452f1d412d : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":5},{"id":"c42b4523ca789a672cc11955ca8c0d12146411e0","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-23 19:24:02.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\nlooking better, some comments inline","accounts_in_message":[],"_revision_number":5},{"id":"dab604f2c45fb0845e7bf9e6469a33eb7f3088e6","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-23 21:42:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"46b2331a48a49a217cd2774266e201f67f30e85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 22:46:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a1f51c47794e77bb3a316c04ebcfb8 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c600c4c7461a4e1a8ebeb96697f7915a : SUCCESS in 5m 59s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/a0d249e02d344fc6b9b6370468dca49d : SUCCESS in 56m 26s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/fee8b37428834563ad6f34df05a78f90 : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":6},{"id":"54147381642def9b99753ce9aba0aa740f71b1eb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-24 12:18:42.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nnice, I left a couple of comments inline and I think should be it from my side\nthanks!","accounts_in_message":[],"_revision_number":6},{"id":"cb69db3d49b874d4bd14481812e62eccb5ba718f","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-24 13:06:24.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"cde4d01035271f943c6093bfc3c55cb231f337fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:56:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3b1c4da02747a7b405f8705eff9337 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecdd5fecaed347cf878c7f90007104f1 : SUCCESS in 4m 59s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/5ba3c733aa5a446b9eed4f70e034efe4 : SUCCESS in 48m 07s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/e1245adc958a4df89272c8564e774db4 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":7},{"id":"716372c03d153cdc46cd71a4bcc07f8a36d9ed12","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-25 15:02:34.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)\n\nlooking better and better, just a couple of final adjustments from my side","accounts_in_message":[],"_revision_number":7},{"id":"c36b3e32e4bf1bbfa22c5ce811e5476b19ad57af","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-25 21:31:48.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"d4b21a21a910b3dc0d477944c4246d5f712ba7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 22:25:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e3e7eed4c54bd8be7a435291fa6e42 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31dbf533e53b49f8b89b4a0f8ed0ae70 : SUCCESS in 5m 07s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/e8f2ead0f8214471a53eecabf727d79b : SUCCESS in 47m 34s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/bed397fc2ef343919a5a016ade3e3312 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":8},{"id":"9c50828aea32a7af0e1014a2ae2dbc720217e257","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-28 09:34:00.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nthanks, tow more formatting changes and it should be good","accounts_in_message":[],"_revision_number":8},{"id":"a9e9eaddfddb8a199986ccdb31bf7def2ef02248","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-28 10:02:01.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"32214f46a7e3e7ef09402ea03b1a01f75e6a1567","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-28 11:11:55.000000000","message":"Patch Set 9: Code-Review+2\n\nthanks khansa :)","accounts_in_message":[],"_revision_number":9},{"id":"66b9bfc7ce34750471dc3f6d796b0b3deb63a577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:12:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ac95e386444f838b456ebf32699399 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962a18c2d734467ca440e7661a254664 : SUCCESS in 9m 25s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d147152a6f8545c6acd3338eae032e8b : SUCCESS in 1h 08m 44s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/eb1afc1107a642358c0d33bb79579c30 : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":9},{"id":"5463826b2be1224b9c66749a6c1445d339177e6d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-28 11:32:13.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d60b8e0103e779a17e7a9e1303e5088046061ac3","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-29 11:42:41.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"e66498981abe53c536ae0ac1acaf780313e5c3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 13:49:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e83c6e3ffaf43078bc7b96e11a5f842 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8d84f3a30f49ffa1efe9d2eeadfdba : SUCCESS in 5m 57s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/b37fef4dd58e4ec99e1df9e21d8f1c9b : SUCCESS in 48m 30s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/0a470581c44d495d9045567746be50b7 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":10},{"id":"d85780b131c85dddaf08314f79d1e873ff7aa714","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-31 09:34:48.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"22c2e3d85d01c9a7af44024cf38a0b461d56b50b","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-31 17:16:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"73de2bf0b84667281ae9e5715fe5abb91af68a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 18:07:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6aadb435ca4e7f94d184e493e9e351 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b18969896644bf78532b21b19727368 : SUCCESS in 6m 30s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9ab6ccb5d4e84e939716b93bc0bed38b : SUCCESS in 47m 39s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/adce0eee2f984d12a6f70492257f8d29 : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":11},{"id":"dc98213f590dba7eaafa6a624f0b0763ca055228","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-31 18:22:34.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"247f6c73427cb2e3bb7a5ebb03ba35fc8a5669b3","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2019-10-31 18:22:54.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"b079b9a9193e10e3cb04d5133030e12da1eb3b4a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-10-31 19:07:53.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"98b69c37b660266d1fb4ada108f8429c046d915e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 19:10:08.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a1cd1fc3a34c8fb7d90d4aaf12d98f : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f330e45f97450f99b6599f0208bc01 : SUCCESS in 5m 51s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9b8bd75c3f5d45408a500b9b22122ba6 : SUCCESS in 44m 09s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/9dfba21bd1db44ea821a6198918b679c : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":12},{"id":"87bccca40d3f3ce083c9863260a95d46305b0e5e","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-31 22:32:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f1e281f8591239ba5c39fc15f07f698a1a73e989","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-31 22:38:03.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9dd1900b3ddb84abca626022d54438c3d7d67e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 23:42:32.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680c9c382e00494eb1f6d81a4a4b8da3 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868722440576401aab132c884fabfa75 : SUCCESS in 7m 09s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/95ef9256a54c4a6d852635e808db5d3a : SUCCESS in 38m 32s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/8a84fd9debaf4569959e54529e83b0fc : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":14},{"id":"c31d9fbd0ad9cbb39fbbd9be65d4656d09aa7d10","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 08:00:47.000000000","message":"Patch Set 14:\n\n(6 comments)","accounts_in_message":[],"_revision_number":14},{"id":"a1198e67dba59b71d833e79eef6a3915ad2e604b","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 11:10:27.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"83f29d14f5a1d28fd708e04a362bb330466dad63","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 11:19:06.000000000","message":"Patch Set 15:\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"bfa0b96e3a0decaa79d9e22f55170be0d96bc05d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 11:19:45.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5b942011f35c6f4a57878b9d39ad3cb227d30860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 12:02:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafb8fa566b442839656e30d56ebb298 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7598427ed07465685afbf314a336239 : SUCCESS in 6m 45s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/2c8439c5fde64f79859166d66e6cf9bc : SUCCESS in 43m 26s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/a9c763f017b94ef28ad2d013da61dd62 : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":15},{"id":"cc0b20f6edc8ca8ac6313e3b84d2211c0c045065","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 13:18:02.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"725aaa89eeebd1b0ba86c75be975be45af4cd65a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 13:41:31.000000000","message":"Patch Set 16:\n\n(6 comments)\n\nI think we are almost there!\n\nBTW, I found this advice useful [1] for me.\n\n1. https://sivers.org/ment","accounts_in_message":[],"_revision_number":16},{"id":"67a7605afb00aec8cd37c823f0f1cc8bc37b2e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 14:04:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add2282704b542709a5fde49a86e9ff5 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9282fe96883944d19a5d28daa6f17af3 : SUCCESS in 6m 03s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d32f5951ea2f4ec7bdcf27fef64c3bc9 : SUCCESS in 44m 10s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/bf801f559d214ed38052419a05fd78b8 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":16},{"id":"ec413d6b37339ca424c9ba8b799447f8d13e91eb","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 15:05:35.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c7564b9c49ea84ad2c879ac382deb5050a9546f4","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 15:10:06.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)\n\nOne tiny nit","accounts_in_message":[],"_revision_number":17},{"id":"44abbfc338d31b65b21a5a5085472cd10574c58b","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 15:14:44.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"18709d5985b9bed0d860f6f9a118bf1828cf051a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 15:49:01.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"1495d68e950b8c2185ade5bd70838731ff0c2a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 15:55:58.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca4cfd4115e44a0b1cfd82425db612d : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427500b4329546b6b3721574462fd09e : SUCCESS in 8m 30s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/aee01e2c551b47248469da7b32c3261a : SUCCESS in 37m 30s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/dfe33e9b88144ced9eb078d336bc7f7d : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":18},{"id":"2f84c58c38b78b9f11a40cb55cf24bff1b51905c","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 16:10:43.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a8b5d0dafe241adc9c1580d2e1ad15ce5ec68e6e","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 16:13:16.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"6af30d643d41032835933d31a96200d3ef4848d1","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 16:15:20.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"2d6ad16764de2d08736a656b66cc706a1ac3f09a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-01 16:17:30.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"223aa777ba7e3ad0901a525127e4ec9f8e200cfa","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-01 16:26:26.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"3e4d90f2cad4859850932760b2bd56626d01c71e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 17:13:22.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b12bbb4b84c402c8f37e7e3855acdb6 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddd56c78b4645b9815d5e2baaf29a94 : SUCCESS in 6m 00s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/73df52a767c24bddaef752ca01a44314 : SUCCESS in 43m 03s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/69734c3183734b22b9e921dd45d5d5ba : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":21},{"id":"d9e76718fdd4e89cc011e474e4e806afc52ce65f","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-04 11:16:59.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"042f4f4955975745d4268bcc7688fa427628a79b","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-07 10:40:46.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"9e0e684ee4a4749f4e00222cc778d49db70c6caa","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-07 11:13:08.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"21f8b555e5880b745fe7f277cbb9fb032d63dffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 11:43:02.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618834d4c3e740b1be2e477151fc209a : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9494c5bde3f14e108fb09c28bcc99212 : SUCCESS in 6m 59s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/96d3cac533b34e58b2526f12cd7b6c8a : SUCCESS in 55m 24s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/1dbd6176bafc4cf58cc9b7eae88f5796 : SUCCESS in 42m 22s","accounts_in_message":[],"_revision_number":22},{"id":"85c6afd03ab9f229b4dd04c2e21eb825c5b3d4d9","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-11 12:56:22.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"646e2084d14b6e50369cb2d329796e6a6d885faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 13:42:28.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2a4b1e14da4e849ca8af79b99f8ec9 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365665683f234fae8dbf84fc9934cffe : SUCCESS in 4m 32s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/60d52d7bf63741ed9b9e4e34b173170f : SUCCESS in 44m 30s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/01e2c44f27fa46649bd9a6b3185c0bde : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":23},{"id":"3a725c3ea7eaff9e9a02e6b43d66f1fa30c480bd","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-12 11:30:53.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"a4ae3c3ccce504a437f0b96ce703a6342b07ce9e","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-18 13:44:11.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"a4c38d0c4721c2bc9c37c469c639d076177e0630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 14:36:59.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a8a8b1ff15f4a1cbd141f06ae76e140 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c0dbebcecd4a52861318c2bce6fb21 : SUCCESS in 7m 45s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/c46ff65aa222405c8854365282725b63 : SUCCESS in 48m 34s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/8c1a228e71e24ff7a3fdefa02c6dca77 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":24},{"id":"18a8c29d04fb6efa842d3172b0eac6953ea44e62","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-25 21:56:03.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"b9e1b353d05cf7d652893a79902f5d43012e1f4a","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-26 09:50:05.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"706dfb0b009036ecf21544f7a72d815e46106057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 10:35:52.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef4e245e4c6474eae602c94a394e931 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f7b7a832d1492a8ab1b30caef07225 : SUCCESS in 5m 23s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/b2e8630d2bca4bdc92e8d7a7d62fb6ab : SUCCESS in 43m 29s","accounts_in_message":[],"_revision_number":25},{"id":"3e7e5f0a19a3d1dbaf1d582e56a6c717586bfc63","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-11-27 10:21:14.000000000","message":"Patch Set 25: Code-Review+1\n\nLooks good to me, let\u0027s see what Ilya thinks before finally approving :)","accounts_in_message":[],"_revision_number":25},{"id":"7db7e78cc76562ae2e14d9e2115008bcc46b9fa5","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-27 12:23:50.000000000","message":"Patch Set 25: Code-Review+1\n\n(2 comments)\n\nLooks almost good to me!","accounts_in_message":[],"_revision_number":25},{"id":"2e215011955f0da9fa5a951ec14a160a0a7fc81e","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-11-28 11:51:08.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"a07074da9cdff14b28eb3093ab850390806f0ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 12:43:47.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489c1250a1854a70a8bf09b8cf21b930 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01f4e998dad423d8b06eb9f622db62e : SUCCESS in 5m 59s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/6f1353284e02402db555fd6bcb9168e5 : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":26},{"id":"e2dafee37eb5954f3c98a06a97c38db49ff4fe74","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-11-29 17:56:52.000000000","message":"Patch Set 26: Code-Review+1\n\n(2 comments)\n\nHopefully, the last nits and we are done! \\o/","accounts_in_message":[],"_revision_number":26},{"id":"19bfe61f3b430502966115ae703fa42523d3f9b8","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-12-01 12:12:50.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"9dc05fd57d8d848d36debce2f4296c6efe92b565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 12:53:43.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ecc0e564cf4f698c412a73ae0cd742 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667d6dde3d2541cba4f2a387be8f924d : SUCCESS in 4m 54s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/ba7707745f5047fc944aeae64ebbd643 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":27},{"id":"abe0a8204c78fdd9d6658d621f9f5b26bf11af73","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-12-01 16:49:28.000000000","message":"Patch Set 27:\n\nNote: it\u0027s convent to mark somehow those review comments you\u0027ve addressed. Otherwise reviewers might consider the patch not being ready for re-review.","accounts_in_message":[],"_revision_number":27},{"id":"c28fa692451d6b331e4d15d44ce780342717fe54","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-12-01 16:49:39.000000000","message":"Patch Set 27: Code-Review+2","accounts_in_message":[],"_revision_number":27},{"id":"e3604cd66f8d3fec47d27dfff2019f0b3cb398c7","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2019-12-01 16:49:44.000000000","message":"Patch Set 27: Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"7b27a4cfe7c27dd425bb12a64a851e24286f50c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 16:49:55.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"1b14688217e832bf98c15d37829662dd6031fab9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 17:29:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":27},{"id":"361f62c6268ea53d1059aad726c0549c42e943fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 17:29:27.000000000","message":"Patch Set 27: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fe9dcdadb5432e8efaf6e65ea6d110 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73122ce7decd4ee8a21232b784b25c1d : SUCCESS in 4m 52s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/fe1f676cd32c4142bb946edd8cc048a9 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":27},{"id":"9893a196be646e6f7c6c6e03d9150f4ed8180054","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 17:30:40.000000000","message":"Patch Set 27:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a500ec3abc64df2ad30175e9714beb6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"015fe9b3b4f28caf26d6051ee2d522f7aa44f826","revisions":{"14288ae07269ebbecb402e0bc499fb9cf12b1855":{"kind":"REWORK","_number":1,"created":"2019-10-21 16:18:48.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/1"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"subject":"update documentation adding Dell EMC Oem extension usage for sushy task number: #36779","message":"update documentation adding Dell EMC Oem extension usage for sushy task number: #36779\n\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/14288ae07269ebbecb402e0bc499fb9cf12b1855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/14288ae07269ebbecb402e0bc499fb9cf12b1855"}]},"branch":"refs/heads/master"},"c736702547217f2c404fdf8bdeddb027d851bcb8":{"kind":"REWORK","_number":2,"created":"2019-10-23 15:37:23.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/2"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-23 15:37:21.000000000","tz":60},"subject":"Story: 2006610 Task: 36779 Change-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6","message":"Story: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c736702547217f2c404fdf8bdeddb027d851bcb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c736702547217f2c404fdf8bdeddb027d851bcb8"}]},"branch":"refs/heads/master"},"fbcf4ca7ab66a15c408b687156a7528e1c2a839d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-23 16:34:10.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/3"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-23 16:33:27.000000000","tz":0},"subject":"Story: 2006610 Task: 36779 Change-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6 Documentation about using OEM Dell EMC extension with sushy it contains a simple script example","message":"Story: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\nDocumentation about using OEM Dell EMC extension with sushy\nit contains a simple script example\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fbcf4ca7ab66a15c408b687156a7528e1c2a839d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fbcf4ca7ab66a15c408b687156a7528e1c2a839d"}]},"branch":"refs/heads/master"},"b759fa8cf019ffecda66b3d8ba3ce29fd3b3e91f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-23 16:58:25.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/4"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-23 16:57:06.000000000","tz":0},"subject":"Using OEM extension Documentation","message":"Using OEM extension Documentation\n\nit contains a simple script example about using\nOEM Dell EMC extension with sushy\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b759fa8cf019ffecda66b3d8ba3ce29fd3b3e91f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b759fa8cf019ffecda66b3d8ba3ce29fd3b3e91f"}]},"branch":"refs/heads/master"},"d074f6bfd7dd0e74ea4490b9b5074c5fd0bd98b8":{"kind":"REWORK","_number":5,"created":"2019-10-23 17:51:28.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/5"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-23 17:50:35.000000000","tz":60},"subject":"Using OEM extension Documentation","message":"Using OEM extension Documentation\n\nIt contains a simple script example about using\nOEM Dell EMC extension with sushy.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d074f6bfd7dd0e74ea4490b9b5074c5fd0bd98b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d074f6bfd7dd0e74ea4490b9b5074c5fd0bd98b8"}]},"branch":"refs/heads/master"},"e16d8ec39098348d682bd3334fc521c2646e1e80":{"kind":"REWORK","_number":6,"created":"2019-10-23 21:42:23.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/6"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-23 21:40:47.000000000","tz":0},"subject":"Using OEM extension Documentation","message":"Using OEM extension Documentation\n\nIt contains a simple script example about using\nOEM Dell EMC extension with sushy.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e16d8ec39098348d682bd3334fc521c2646e1e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e16d8ec39098348d682bd3334fc521c2646e1e80"}]},"branch":"refs/heads/master"},"f0ad44cb85243d148b3028cddc82b6f7a43ec8f2":{"kind":"REWORK","_number":7,"created":"2019-10-24 13:06:24.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/7"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-24 13:06:04.000000000","tz":0},"subject":"Using OEM extension Documentation","message":"Using OEM extension Documentation\n\nIt contains a simple script example about using\nOEM Dell EMC extension with sushy.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f0ad44cb85243d148b3028cddc82b6f7a43ec8f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f0ad44cb85243d148b3028cddc82b6f7a43ec8f2"}]},"branch":"refs/heads/master"},"42cd55b91deaa693bd9153f410e6c321e0d853bb":{"kind":"REWORK","_number":8,"created":"2019-10-25 21:31:48.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/8"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-25 21:31:39.000000000","tz":0},"subject":"Using OEM extension Documentation","message":"Using OEM extension Documentation\n\nIt contains a simple script example about using\nOEM Dell EMC extension with sushy.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/42cd55b91deaa693bd9153f410e6c321e0d853bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/42cd55b91deaa693bd9153f410e6c321e0d853bb"}]},"branch":"refs/heads/master"},"5c5f718853c73cc4ac29f413c3336dea8c69022b":{"kind":"REWORK","_number":9,"created":"2019-10-28 10:02:01.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/9"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-28 10:01:44.000000000","tz":0},"subject":"Using OEM extension Documentation","message":"Using OEM extension Documentation\n\nIt contains a simple script example about using\nOEM Dell EMC extension with sushy.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5c5f718853c73cc4ac29f413c3336dea8c69022b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5c5f718853c73cc4ac29f413c3336dea8c69022b"}]},"branch":"refs/heads/master"},"a0a6244c320e1d7c3cf5b7b911a51ae5e32a3902":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-10-29 11:42:41.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/10"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-29 11:17:52.000000000","tz":0},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating \nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a0a6244c320e1d7c3cf5b7b911a51ae5e32a3902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a0a6244c320e1d7c3cf5b7b911a51ae5e32a3902"}]},"branch":"refs/heads/master"},"f375af2f3a6d691ba71c605f0393e7470af30f7c":{"kind":"REWORK","_number":11,"created":"2019-10-31 17:16:20.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/11"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-31 17:15:49.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f375af2f3a6d691ba71c605f0393e7470af30f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f375af2f3a6d691ba71c605f0393e7470af30f7c"}]},"branch":"refs/heads/master"},"5a01bea4e0830cf13b1a120824cb88dd891322cb":{"kind":"REWORK","_number":12,"created":"2019-10-31 18:22:34.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/689835/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/12"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2019-10-31 18:22:32.000000000","tz":0},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5a01bea4e0830cf13b1a120824cb88dd891322cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5a01bea4e0830cf13b1a120824cb88dd891322cb"}]},"branch":"refs/heads/master"},"d6b4f472b9ff646b4d441a4f82df33ac82c03909":{"kind":"REWORK","_number":13,"created":"2019-10-31 22:32:26.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/13"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-31 22:32:06.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d6b4f472b9ff646b4d441a4f82df33ac82c03909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d6b4f472b9ff646b4d441a4f82df33ac82c03909"}]},"branch":"refs/heads/master"},"d02c7fddaee23cb3222c64c393bc68c4651bd682":{"kind":"REWORK","_number":14,"created":"2019-10-31 22:38:03.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/14"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-31 22:35:59.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d02c7fddaee23cb3222c64c393bc68c4651bd682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d02c7fddaee23cb3222c64c393bc68c4651bd682"}]},"branch":"refs/heads/master"},"e206bc2fe8586712146c78f19a2bf5ab1dd90428":{"kind":"REWORK","_number":15,"created":"2019-11-01 11:10:27.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/15"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 11:09:48.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e206bc2fe8586712146c78f19a2bf5ab1dd90428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e206bc2fe8586712146c78f19a2bf5ab1dd90428"}]},"branch":"refs/heads/master"},"94192bba2dcaaab0fa10584bcd7b651b3687d469":{"kind":"REWORK","_number":16,"created":"2019-11-01 13:18:02.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/16"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 13:17:25.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/94192bba2dcaaab0fa10584bcd7b651b3687d469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/94192bba2dcaaab0fa10584bcd7b651b3687d469"}]},"branch":"refs/heads/master"},"f9c74e875f191926237621bec4ded9947f743364":{"kind":"REWORK","_number":17,"created":"2019-11-01 15:05:35.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/17"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 15:04:44.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f9c74e875f191926237621bec4ded9947f743364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f9c74e875f191926237621bec4ded9947f743364"}]},"branch":"refs/heads/master"},"5fe5a3ae36d5eb90f69404c74d9c418a7290d39d":{"kind":"REWORK","_number":18,"created":"2019-11-01 15:14:44.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/18"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 15:14:33.000000000","tz":0},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5fe5a3ae36d5eb90f69404c74d9c418a7290d39d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5fe5a3ae36d5eb90f69404c74d9c418a7290d39d"}]},"branch":"refs/heads/master"},"5bbcd102bfed8be6889918339e988a3906aa3c8a":{"kind":"REWORK","_number":19,"created":"2019-11-01 16:10:43.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/19"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 16:10:11.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5bbcd102bfed8be6889918339e988a3906aa3c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5bbcd102bfed8be6889918339e988a3906aa3c8a"}]},"branch":"refs/heads/master"},"e145c312aa0d74c0e717af45499b8a95cce65ec8":{"kind":"REWORK","_number":20,"created":"2019-11-01 16:15:20.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/20"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 16:15:09.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e145c312aa0d74c0e717af45499b8a95cce65ec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e145c312aa0d74c0e717af45499b8a95cce65ec8"}]},"branch":"refs/heads/master"},"d7e84ade54ed135f6c4ade459047bb50af3713fb":{"kind":"REWORK","_number":21,"created":"2019-11-01 16:26:26.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/21"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-01 16:25:57.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d7e84ade54ed135f6c4ade459047bb50af3713fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d7e84ade54ed135f6c4ade459047bb50af3713fb"}]},"branch":"refs/heads/master"},"57700501de38af22486649c6a3cdc3caf0eed347":{"kind":"REWORK","_number":22,"created":"2019-11-07 10:40:46.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/22"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-07 10:40:29.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/57700501de38af22486649c6a3cdc3caf0eed347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/57700501de38af22486649c6a3cdc3caf0eed347"}]},"branch":"refs/heads/master"},"9588b97cc8263ebb9db5e41c1537f819d5868fe8":{"kind":"REWORK","_number":23,"created":"2019-11-11 12:56:22.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/23"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-11 12:55:48.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9588b97cc8263ebb9db5e41c1537f819d5868fe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9588b97cc8263ebb9db5e41c1537f819d5868fe8"}]},"branch":"refs/heads/master"},"09eee0a6fe518bf52bd4ef8965dc269956ea3ba5":{"kind":"REWORK","_number":24,"created":"2019-11-18 13:44:11.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/24"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-18 13:43:31.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/09eee0a6fe518bf52bd4ef8965dc269956ea3ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/09eee0a6fe518bf52bd4ef8965dc269956ea3ba5"}]},"branch":"refs/heads/master"},"175b3dc575d59e6087b40e51ac26da2af7128f96":{"kind":"REWORK","_number":25,"created":"2019-11-26 09:50:05.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/25"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-26 09:49:07.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/175b3dc575d59e6087b40e51ac26da2af7128f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/175b3dc575d59e6087b40e51ac26da2af7128f96"}]},"branch":"refs/heads/master"},"0ce74b2ff1eeae73a971eed2180bc10ccee359ea":{"kind":"REWORK","_number":26,"created":"2019-11-28 11:51:08.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/26"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-11-28 11:49:51.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0ce74b2ff1eeae73a971eed2180bc10ccee359ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0ce74b2ff1eeae73a971eed2180bc10ccee359ea"}]},"branch":"refs/heads/master"},"015fe9b3b4f28caf26d6051ee2d522f7aa44f826":{"kind":"REWORK","_number":27,"created":"2019-12-01 12:12:50.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/35/689835/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/689835/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/689835/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/689835/27"}}},"commit":{"parents":[{"commit":"33879254e3fb04c531e811b2f4dd37b06e6c828a","subject":"Fix typo in the section Enabling SSL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33879254e3fb04c531e811b2f4dd37b06e6c828a"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 16:18:32.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-12-01 12:12:17.000000000","tz":60},"subject":"Add OEM extension example script","message":"Add OEM extension example script\n\nThis change adds a simple example script illustrating\nOEM extension call based on Acme sushy extension package.\n\nStory: 2006610\nTask: 36779\nChange-Id: Icf22b9d38e307a07f59112a6c462d48a8ec6c3d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/015fe9b3b4f28caf26d6051ee2d522f7aa44f826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/015fe9b3b4f28caf26d6051ee2d522f7aa44f826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
