)]}'
{"id":"openstack%2Fsushy~689849","triplet_id":"openstack%2Fsushy~master~I1bacb6c9f832c2a33e13952726a4451e35362fb9","project":"openstack/sushy","branch":"master","topic":"oem_extension_documentation","hashtags":[],"change_id":"I1bacb6c9f832c2a33e13952726a4451e35362fb9","subject":"update sushy usage documentation specifying version range for OEM extension","status":"ABANDONED","created":"2019-10-21 17:26:28.000000000","updated":"2019-10-23 13:20:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9625aa5a40299fc509f7ac75614fb24e106caa11","_number":689849,"virtual_id_number":689849,"owner":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-21 18:25:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 18:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5509a5881723c6264b9c91d84136ccaf8d2e6368","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-21 17:26:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"541683c4ede5dd4ab62096089acf08ce69295070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 18:25:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad6fb97f8384b38a575a2b9816a8673 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f092436a82de4b2bb5a3bc3e0f131f39 : FAILURE in 7m 25s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/4c68e411e0ec409390ecbc296df7ff5c : SUCCESS in 52m 11s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/fbf585731d6841c5a1bc1056bfd7559c : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"35f50de6648685b12b15adad3b7f2e8781529f5f","author":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"date":"2019-10-23 13:20:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96edf14b1c0dac76965193ddf2219d294b9c3db0","revisions":{"96edf14b1c0dac76965193ddf2219d294b9c3db0":{"kind":"REWORK","_number":1,"created":"2019-10-21 17:26:28.000000000","uploader":{"_account_id":31182,"name":"khansa","email":"khansa.amrouni@gmail.com","username":"khansa"},"ref":"refs/changes/49/689849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/689849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/689849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/689849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/689849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/689849/1"}}},"commit":{"parents":[{"commit":"35c2156fb9dc7dc6fa0b43d1ed434d98dc32a8c0","subject":"update sushy usage documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/35c2156fb9dc7dc6fa0b43d1ed434d98dc32a8c0"}]}],"author":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 17:24:41.000000000","tz":60},"committer":{"name":"khansa","email":"khansa.amrouni@gmail.com","date":"2019-10-21 17:24:41.000000000","tz":60},"subject":"update sushy usage documentation specifying version range for OEM extension","message":"update sushy usage documentation specifying version range for OEM extension\n\nChange-Id: I1bacb6c9f832c2a33e13952726a4451e35362fb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/96edf14b1c0dac76965193ddf2219d294b9c3db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/96edf14b1c0dac76965193ddf2219d294b9c3db0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
