)]}'
{"id":"openstack%2Fsushy~700403","triplet_id":"openstack%2Fsushy~master~I73f57c423dfacdb67d9e06de2885943d800f47a9","project":"openstack/sushy","branch":"master","topic":"tox-enforce-correct-pyver","hashtags":[],"change_id":"I73f57c423dfacdb67d9e06de2885943d800f47a9","subject":"Enforce running tox with correct python version based on env","status":"MERGED","created":"2019-12-23 09:46:33.000000000","updated":"2020-01-09 12:34:01.000000000","submitted":"2020-01-09 12:31:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700403-1578573077575-1b68ee4e","meta_rev_id":"c7f9af4835ac9201841a3f709ea3708a65e3591a","_number":700403,"virtual_id_number":700403,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 12:31:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-01-09 10:56:56.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-26 11:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2020-01-09 10:56:56.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-26 11:41:06.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-01-09 10:56:56.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-01-09 12:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8801a61e5ae7fd627fe3c92fdaa2d3c9a1dcc954","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-23 09:46:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"950da4470c3d30deebb5e9a1d0709eda1bb37863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 10:31:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69d98d11db234a69b7f3385a5cb6ac9c : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0243be2595b4f13954d12ebdce93767 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8325f91653f4b4f96d4c702ade0564a : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b95786442c38419c989b986c647bd556 : SUCCESS in 2m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc5f545652f54f08b58259a4da2d5828 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5ca292f1354605abcc8d7bf50e05e4 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ba6bfbc3c4b43f49334cf50a4b3d21e : SUCCESS in 3m 18s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/66ff3556088e40879cdc6857b5a76e3f : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":1},{"id":"78f5025f6a5de0595a750902cbb92d785f75f4cc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-12-26 11:41:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"52926df63fa49457e8072b1bdb4e3d6f08f50ae5","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-01-09 10:56:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you Riccardo.","accounts_in_message":[],"_revision_number":1},{"id":"45682af2bd312c2b2a7bd2a1226b3a8af2237f1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:57:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64ef9694d673b6f7d64ffe76311dda526a71b87b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:31:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d25d4f4fb0afb05cf01fae7e9a34d499c8bee6b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:31:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc5b69eb35ec4e5c86ed8ca7fc167f67 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e293d343f7f447da4d110106e1ce5c9 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa42c8599d440f5a55b8f5d50d9ba10 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24485513dbb74cdab5844849bbe82e52 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbd469986b8468d87b398ccc70fa31e : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3fae4b2b8084b2c8c3c50b8658a21f2 : SUCCESS in 4m 26s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/c89a4c1f4c7b41c9854fbd83f8c38a41 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e57a29cbfd44e4a2dcbe206b710777af1f4d7e1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f8fae6b7d44ab6b5578f6e7270a69d : SUCCESS in 2m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/875c97e25b064970bb4aecec11ce2a72 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69","revisions":{"c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69":{"kind":"REWORK","_number":1,"created":"2019-12-23 09:46:33.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/03/700403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/03/700403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/03/700403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/03/700403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/03/700403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/03/700403/1"}}},"commit":{"parents":[{"commit":"9d01756833ec97a05f301a4fa1e8b2511d7c5c74","subject":"SSC.disks_sizes_bytes handle CapacityBytes is None","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9d01756833ec97a05f301a4fa1e8b2511d7c5c74"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-23 09:46:14.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-12-23 09:46:14.000000000","tz":60},"subject":"Enforce running tox with correct python version based on env","message":"Enforce running tox with correct python version based on env\n\nSince removing support for Python 2, we changed the basepython\nvalue to 3.\nThis means that all the tox tests run with the default python\nversion available in the system.\nThis is not quite correct when running on environment such as\npy36, py37 or py38, since they imply running with different\nPython versions based on the environment.\nTo enforce the correct version we need to add the option\nignore_basepython_conflict available since tox 3.1.0 [0].\n\n[0] https://tox.readthedocs.io/en/latest/config.html#conf-ignore_basepython_conflict\n\nChange-Id: I73f57c423dfacdb67d9e06de2885943d800f47a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
