)]}'
{"id":"openstack%2Fsushy~703652","triplet_id":"openstack%2Fsushy~master~I8d1e1691a0bb2b6315894c85569a73633d34c1cb","project":"openstack/sushy","branch":"master","topic":"11-fix-none-literal","hashtags":[],"change_id":"I8d1e1691a0bb2b6315894c85569a73633d34c1cb","subject":"Fix \u0027None\u0027 field value processing","status":"MERGED","created":"2020-01-21 16:31:44.000000000","updated":"2020-02-13 00:06:33.000000000","submitted":"2020-02-10 18:03:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703652-1581357785319-bbef5f2b","meta_rev_id":"dba2bb745fb65c9ce54452e25f4f62530b45cf79","_number":703652,"virtual_id_number":703652,"owner":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"date":"2020-02-11 16:46:00.000000000","post_submit":true,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 18:03:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":2,"date":"2020-02-10 17:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-02-10 17:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-10 17:20:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-02-10 17:07:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 09:02:41.000000000","updated_by":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"reviewer":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"state":"REVIEWER"},{"updated":"2020-01-30 21:16:59.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-02-03 14:40:14.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-02-10 17:05:11.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-02-10 17:07:01.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-02-10 17:20:12.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-02-10 18:03:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-11 16:46:00.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"}],"messages":[{"id":"911c4889c5cbf44bdf8c3d8092505ea15c9d8b16","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-21 16:31:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c0e442d8a6fb92a2fb20c03a6c625d5abfd2a4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 17:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0f228fbdc0245c6952e4ad938ea169c : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ef492be17bd4d7e8c5cb784c2675ce4 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f613abeb7154233834ea2c8c081f8be : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/496d48fb535c4369a12a05d5f2d2cb51 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12bf247478d743e1b110d0174454de00 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183766b4fc1e441a86fc6fd775adb48a : SUCCESS in 6m 29s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/7f9a3fe07ddd4d84abcdb6ed9dbbadb7 : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fe66fbcd9432819aef9bb4f3beac13e9419a9eba","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-21 17:58:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-21:10:31/703652/1/ : SUCCESS in 1h 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"59d50415dba15cf537d6306101a9b56ebf6a9538","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-01-21 20:57:19.000000000","message":"Patch Set 1:\n\ndell-recheck","accounts_in_message":[],"_revision_number":1},{"id":"8b2562f9a3436284b70f58a4ee62de096f0590ba","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-21 21:38:21.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-21:14:57/703652/1/ : FAILURE in 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"36b6bf4ae1126d436cf7382f44463148865e120e","author":{"_account_id":6133,"name":"yolanda.robla","email":"yroblamo@redhat.com","username":"yolanda.robla"},"date":"2020-01-22 09:02:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9c374c9a28ed98500b0a4fa1742617370472a0e9","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-22 14:36:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"36ebddeae86008bf97f5c112ac421d4d02ba5fbe","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-22 20:10:21.000000000","message":"Patch Set 1:\n\ndell-recheck","accounts_in_message":[],"_revision_number":1},{"id":"b141f4a0d578ac3e6beb57392b3f71a50eae01bc","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-01-22 20:10:36.000000000","message":"Patch Set 1: Workflow-1\n\nHolding until we see dell ci happiness.","accounts_in_message":[],"_revision_number":1},{"id":"b67b14a306b15485726853437d912079132111ca","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-22 20:23:48.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-22:14:10/703652/1/ : ABORTED in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6af27377eff6b0ced3564f1961650fc2485c61ca","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-23 08:35:46.000000000","message":"Patch Set 1:\n\ndell-recheck aborted?!?","accounts_in_message":[],"_revision_number":1},{"id":"9b575d9350d81b34219fbcbb4da3d17a874a71e1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-01-24 08:59:53.000000000","message":"Patch Set 1:\n\ndell-recheck","accounts_in_message":[],"_revision_number":1},{"id":"bfeae28f6f3a041ef9c558c3146ee121303c9869","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-01-27 18:25:00.000000000","message":"Patch Set 1:\n\ndell-recheck","accounts_in_message":[],"_revision_number":1},{"id":"5d864b1ea90bcf7d7887bcdbc47f648957937dc6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-27 20:25:08.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-27:12:25/703652/1/ : FAILURE in 1h 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7ac87882536e2869f8b806f1f69b0e75c161672a","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-28 05:42:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"877ebda2c37da9089a8b46fcff0881ed954217cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 06:34:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05269c44e0cc4c8288d0bc8db183087b : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c060b9b686745ef8652a7e0441fc8b3 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ae7a1b832c46d38ea26f1e501b0c9a : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8868d580800492ead1782424cc7f146 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52d2df6d48394b15b141f6b7a11cec35 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7a41265e49407aa33cad0b1f53554d : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb75f7cdf5e24ad29b50d3e4d553f33b : SUCCESS in 5m 18s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/60382e9ecd6f4ca981e72d65fd54cb46 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":2},{"id":"5bea2b6f276c02daaf6dbe7222c1bd03850f9e5d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-28 07:30:38.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-27:23:42/703652/2/ : FAILURE in 1h 46m 50s","accounts_in_message":[],"_revision_number":2},{"id":"bb95efff75d27ce68595887b8cfa36ce15ea567d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-28 11:14:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36d4d78f1825d7651e915441c747af4de3a983e1","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-28 11:21:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"47c07858360495e15683ada4908c9d7a6e8a2f92","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-28 12:28:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cfa5e9e3e98809537e14e7d92f33261cf0cfd474","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-28 12:36:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94febbb816606cd326f7308ab537e77f2d25ef24","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-28 12:41:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"22f10550a2017d5a900cb6cd15e0b3e754494674","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-01-28 13:00:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e5715dd7466f1b63a9bb805fd05fe35eb3b9f66","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-01-28 14:20:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94f133f06f433aa14c6ed54b1a0c29bcedb92aaa","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-01-30 21:16:59.000000000","message":"Patch Set 2:\n\ndell-recheck","accounts_in_message":[],"_revision_number":2},{"id":"1a4fbeef8c8a31c0766ba05d4f5d64402463ef6e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-30 23:04:01.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-30:15:17/703652/2/ : FAILURE in 1h 46m 51s","accounts_in_message":[],"_revision_number":2},{"id":"eda28cb7bd1d964a8dd4c8cd2b91efa7b7a43c88","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-31 16:20:44.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-31:09:55/703652/2/ : ABORTED in 24m 14s","accounts_in_message":[],"_revision_number":2},{"id":"51c17a43d8ae756d04075aa805a343f5cd14098b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-01-31 17:04:10.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-01-31:10:21/703652/2/ : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b3f25f15ded5500c3be6bf1eb0728fd58fda0038","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-02-03 14:40:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"24255032cafed3a5137bd1ac3f9b41b0154a9e86","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-03 14:50:00.000000000","message":"Patch Set 2: Code-Review+1\n\nI\u0027m leaving a +1 considering that the patch is actually very needed, but the discussion is still on-going, and I actually have the same concern about handling possible null values","accounts_in_message":[],"_revision_number":2},{"id":"a643de4010282b56ebd249641fc0589d0101232d","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-02-04 11:03:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"359298aacef16fbc35b9e24df93a31479779cc8c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-02-05 00:04:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5069715e3ac1f173c6db96040e085fb862a147f7","author":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"date":"2020-02-10 15:27:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a120b8d3542d3f988ecbe68675a33dbc307e74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:27:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f62950dce33347838e5576ed75fb9416 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17a3ab7955e442c1a1be2f966c326543 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378d88d21d4645cbab4f5e493e875279 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f8adafa4c7e489ab23250f6082769e8 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c7453a87721470ea1d68a2e95d8354e : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cef62f319ff64b61af7ad653e28d170c : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea6cf3079d447619156d7fc5a17b94f : SUCCESS in 10m 23s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/3ca394f20e8c47c4a46c145de8c2bcf0 : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9b87774817e68bbef5b4c3f668f4bd7ca80ec76d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-02-10 17:05:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d216b743858cf218b4a7a447ba6982fe84072396","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-02-10 17:07:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7d926539dff02c7385c9cf2e90034fb975aa1087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:07:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c8ef4a135a69ccddf6f58c4536d3f4537beab66b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-02-10 17:20:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9dccc036b695a558a9d671f1a6833c49cad76457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:03:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"17f1221e58378759762f4c51732023d5bf106608","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:03:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f6071a81eaf43eeabb0f01252cfc0d0 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69069c33e51547fd9e4e9fd435b8e956 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a70199a73d341aa8fa840ec7d413ce4 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1c1e876081c45f4bcc41de9098efcc4 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc84fd9f98da44c6aef736a84f2c268b : SUCCESS in 8m 36s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/882c6e26ae1f46bdbe9bed951a764d80 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":3},{"id":"796504e1885629364c0392650f0f6809016fa578","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:07:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa8b3f8678148a4b59b55320a8d9f30 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"30447df8472945372fddfe8316933fa7756ffdd1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-02-11 16:46:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-02-10:09:27/703652/3/ : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":3},{"id":"153ea04534390663efa283059097f2e611b98184","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-02-13 00:06:33.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ad426df30560ab202f0141f58875b4b0b3c392ed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1ec13aaba2d0b9c703a2564ec588e4d6de92697b","revisions":{"d840293aa6d597513773748e5a1491ae011d55ee":{"kind":"REWORK","_number":1,"created":"2020-01-21 16:31:44.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/52/703652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/52/703652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/52/703652/1"}}},"commit":{"parents":[{"commit":"c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69","subject":"Enforce running tox with correct python version based on env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c944ab44d18fafa8dbd74bdf45ee4d2c93b9da69"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-21 16:31:11.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-21 16:31:11.000000000","tz":60},"subject":"Fix \u0027None\u0027 field value processing","message":"Fix \u0027None\u0027 field value processing\n\nChange [1] introduced a conditional call to Field\u0027s\n\"adapter\" callable supposedly to work-around some OEM issues.\nSadly, the exact problem that change [1] has been trying to\nsolve was never explained and remains murky.\n\nHowever, fix [1] has also introduced a bug which makes sushy\nignoring \u0027None\u0027 literals in the values. This sometimes leads\nto sushy failing to operate on perfectly valid JSON documents.\n\nThis change removes all conditions guarding Field\u0027s \"adapter\"\nfunction calls.\n\n1. https://review.opendev.org/#/c/669963/6\n\nChange-Id: I8d1e1691a0bb2b6315894c85569a73633d34c1cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d840293aa6d597513773748e5a1491ae011d55ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d840293aa6d597513773748e5a1491ae011d55ee"}]},"branch":"refs/heads/master"},"6e1a9f3331891bf54ec85615797f85b2af697769":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-28 05:42:20.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/52/703652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/52/703652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/52/703652/2"}}},"commit":{"parents":[{"commit":"544f000fe68fd214b0db553011c0a0862790a05f","subject":"Merge \"Add Task Monitor support for async operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/544f000fe68fd214b0db553011c0a0862790a05f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-21 16:31:11.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-28 05:42:07.000000000","tz":60},"subject":"Fix \u0027None\u0027 field value processing","message":"Fix \u0027None\u0027 field value processing\n\nChange [1] introduced a conditional call to Field\u0027s\n\"adapter\" callable supposedly to work-around some OEM issues.\nSadly, the exact problem that change [1] has been trying to\nsolve was never explained and remains murky.\n\nHowever, fix [1] has also introduced a bug which makes sushy\nignoring \u0027None\u0027 literals in the values. This sometimes leads\nto sushy failing to operate on perfectly valid JSON documents.\n\nThis change removes all conditions guarding Field\u0027s \"adapter\"\nfunction calls.\n\n1. https://review.opendev.org/#/c/669963/6\n\nChange-Id: I8d1e1691a0bb2b6315894c85569a73633d34c1cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e1a9f3331891bf54ec85615797f85b2af697769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e1a9f3331891bf54ec85615797f85b2af697769"}]},"branch":"refs/heads/master"},"1ec13aaba2d0b9c703a2564ec588e4d6de92697b":{"kind":"REWORK","_number":3,"created":"2020-02-10 15:27:35.000000000","uploader":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"ref":"refs/changes/52/703652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/52/703652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/703652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/52/703652/3"}}},"commit":{"parents":[{"commit":"544f000fe68fd214b0db553011c0a0862790a05f","subject":"Merge \"Add Task Monitor support for async operations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/544f000fe68fd214b0db553011c0a0862790a05f"}]}],"author":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-01-21 16:31:11.000000000","tz":60},"committer":{"name":"Ilya Etingof","email":"etingof@gmail.com","date":"2020-02-10 15:27:21.000000000","tz":60},"subject":"Fix \u0027None\u0027 field value processing","message":"Fix \u0027None\u0027 field value processing\n\nChange [1] introduced a conditional call to Field\u0027s\n\"adapter\" callable supposedly to work-around some OEM issues.\nSadly, the exact problem that change [1] has been trying to\nsolve was never explained and remains murky.\n\nHowever, fix [1] has also introduced a bug which makes sushy\nignoring \u0027None\u0027 literals in the values. This sometimes leads\nto sushy failing to operate on perfectly valid JSON documents.\n\nThis change removes most of the conditions guarding Field\u0027s\n\"adapter\" function calls.\n\n1. https://review.opendev.org/#/c/669963/6\n\nChange-Id: I8d1e1691a0bb2b6315894c85569a73633d34c1cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1ec13aaba2d0b9c703a2564ec588e4d6de92697b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1ec13aaba2d0b9c703a2564ec588e4d6de92697b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
