)]}'
{"id":"openstack%2Fsushy~718617","triplet_id":"openstack%2Fsushy~master~I1e3dabe9c6e37127cfba54940fbf1a832447bb6a","project":"openstack/sushy","branch":"master","topic":"ironic-convert-dib","hashtags":[],"change_id":"I1e3dabe9c6e37127cfba54940fbf1a832447bb6a","subject":"Convert sushy tempest to dib","status":"MERGED","created":"2020-04-09 08:44:45.000000000","updated":"2020-04-13 09:58:59.000000000","submitted":"2020-04-13 09:55:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718617-1586771728323-eaa21e4c","meta_rev_id":"e8e1e6c0c77fab318164ab8909c9e418be4d3d1c","_number":718617,"virtual_id_number":718617,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2020-04-10 02:38:37.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 09:55:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-04-13 08:48:15.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-12 15:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2020-04-13 08:48:15.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 02:38:37.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-04-12 15:24:46.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-13 08:48:15.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-04-13 09:55:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6768a812d86fdf9e0ce50b896a87349e52b6ac52","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 08:44:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5473c720d473090eaadad0c26c7a72b71533e3aa","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 08:45:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7aedc6f9a3728ed9f4981e8352d79105da4ae3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:44:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee43aa583ef40739dbdad5334f1f50f : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/392522beddd44217ba10a2653eeb1401 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c40119f4723425a86afcf2f9dacc9ce : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87f0cfd1a1564c8ba33aaa429f2c66dc : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/894f435fe95e4c78a0ba49300c8897fd : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30b59f28159840048bb9631df5dd392d : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceaeca1a78524d0698afa4d589968dd9 : SUCCESS in 10m 15s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/43be557037854178b9fa70144cd28ee4 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":2},{"id":"d480cf53b3bc9ac1b91ab39ede58848a65de51b6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-10 02:38:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-09:03:45/718617/2/ : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4bbdbc06f8647feaff1207839cd89fad1f739c05","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-12 15:24:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"79f3891ae7cfdf9024db3a00eae2786886313818","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-04-13 08:48:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you Riccardo.","accounts_in_message":[],"_revision_number":2},{"id":"79bd142d0c9762ec14b37654f6fc67a058f43933","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 08:48:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76d4911a2566b59419f3e12b83dbfc149a9f6c21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:55:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334e50c188aa4d34957f656851bf0213 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7c585811ba46ab84bbbb70ab8af3dd : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b130a204d2d941e7a3d75482138564b4 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6bdf5b5ddff4fd6bc2947321a9edc74 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730626ff4ee942bebf18c5a4f6dc5daa : SUCCESS in 6m 31s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/6893d515b99744b2a486b77ed4bbfc33 : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":2},{"id":"32c3c6108696068f19e1097682bb20f58f0e5a5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:55:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"731fd9e5654581132b824a3e14c36609e27c2ce6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:58:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f80126473140dd90dd2ccd5f68e605 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"226781af6143c28bb5b0acabfb37737569794f53","revisions":{"3e3df7240950bf2ea546f25e1fcfa3de51e0a00e":{"kind":"REWORK","_number":1,"created":"2020-04-09 08:44:45.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/17/718617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/17/718617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/17/718617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/17/718617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/17/718617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/17/718617/1"}}},"commit":{"parents":[{"commit":"9eae135fe1f23ef99c132b82af7983ec1da275da","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9eae135fe1f23ef99c132b82af7983ec1da275da"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 08:44:28.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 08:44:28.000000000","tz":120},"subject":"Convert ironicclient-tempest to dib","message":"Convert ironicclient-tempest to dib\n\nDepends-On: https://review.opendev.org/718607\nChange-Id: I1e3dabe9c6e37127cfba54940fbf1a832447bb6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3e3df7240950bf2ea546f25e1fcfa3de51e0a00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3e3df7240950bf2ea546f25e1fcfa3de51e0a00e"}]},"branch":"refs/heads/master"},"226781af6143c28bb5b0acabfb37737569794f53":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-09 08:45:16.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/17/718617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/17/718617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/17/718617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/17/718617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/17/718617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/17/718617/2"}}},"commit":{"parents":[{"commit":"9eae135fe1f23ef99c132b82af7983ec1da275da","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9eae135fe1f23ef99c132b82af7983ec1da275da"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 08:44:28.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-09 08:44:58.000000000","tz":120},"subject":"Convert sushy tempest to dib","message":"Convert sushy tempest to dib\n\nDepends-On: https://review.opendev.org/718607\nChange-Id: I1e3dabe9c6e37127cfba54940fbf1a832447bb6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/226781af6143c28bb5b0acabfb37737569794f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/226781af6143c28bb5b0acabfb37737569794f53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
