)]}'
{"id":"openstack%2Fsushy~720795","triplet_id":"openstack%2Fsushy~master~Iace21f058714365115e2cd63c3a463af27540721","project":"openstack/sushy","branch":"master","topic":"lower-constraints","hashtags":[],"change_id":"Iace21f058714365115e2cd63c3a463af27540721","subject":"Stop configuring install_command in tox.","status":"MERGED","created":"2020-04-17 18:14:21.000000000","updated":"2020-04-28 15:07:41.000000000","submitted":"2020-04-28 15:04:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720795-1588086254710-81340386","meta_rev_id":"b97e92bebb0076e3ed169c4acd58bee52f1dd32d","_number":720795,"virtual_id_number":720795,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-04-28 01:56:00.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 15:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-27 18:43:02.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-04-20 08:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-04-28 14:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2020-04-28 14:10:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 08:23:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-04-28 01:56:00.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-04-28 14:10:40.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-28 15:04:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02570cbd4a898e2c6078402243f78da43976984e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-17 18:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ede941aa723b3aab39b70d945f53ad28143f15de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 19:28:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49265423f6904fbc8fcd972f177abfc7 : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b3ab9e70f464034aa245d89363d932c : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2fa2ba985b4e48ababf99ec6a9e09b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e398a321c23453a80b7c10211222a06 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d502fd51ce314f97a4777eff06d5fc49 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b008382851407a8e1b85dfcda6807a : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/743b22c9edba41799d055d1323513df8 : SUCCESS in 3m 52s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/b241deda3fb7420b9186029922619489 : FAILURE in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"41c168546f9f7caaa7fef6744622757c624a0ac9","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-17 19:29:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60e099774dbbc9cd12b3c95cd37a6db40f97e7f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:25:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d097edfefe2452686449a40531e28cf : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8df5bff8a6714ba9a281dd2e57dbab3c : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16023e97bb344af6ace2cf637bcef3fb : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2db0db7eb251457eb09bbb63d06e5244 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91ce4af5eb374ab59915cb688f6ad4cb : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff14d7f712f44d5ea040c886f80e2a12 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e86603621534af094f083fda8b84f37 : SUCCESS in 3m 39s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/c9588790fd20481bb1c72ca1d0b051da : FAILURE in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a8c909bc173f322e5c7130a7682b796382f8e588","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-18 07:42:44.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-17:13:14/720795/1/ : FAILURE in 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6df95d044fe6129f6c15f892840d1106d0b26089","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-20 08:23:03.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35373fa7c1de1d06fc53381316888931d03158be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 09:17:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f73482a9af4062abd3bc2816611dfc : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f332a1f04a64528b0109efc10743bab : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c88dc7617f4128bd7f83d7d2847fd6 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef0f2724c1f94c15900d4100f1035765 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810cdfe96f5348e59658d326427bb378 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4362af05afc5443a8d4fd1afeaa7d924 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f59319188aa4e80bf39fff5a9a07383 : SUCCESS in 5m 15s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/8f648bf53abd437c93fe1ea8101045a2 : FAILURE in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d98e85eec74b0b21b5569f245db69e47a13def4d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-20 11:50:01.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-20:03:22/720795/1/ : FAILURE in 42m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8675c096c26b82175cb1297e82468903e6579f98","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-20 12:26:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"05a220541e945183d66aa767b540738c72a22cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 16:40:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41c6d65cf2f94b8e8b4941f47c291b25 : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f882dd03a87a49d69b4e44d16effa4ad : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69befd432f2489f959a4845f79eb392 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55c0a8b6bf3a4176af4caee95d7e645b : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec335895042c4d72aca1eb45da76eb35 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4eb835194514e6cb5e2d2a961099bf8 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0c43d9da7e94ec2a5a5b25e778a0918 : SUCCESS in 4m 02s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/aef934dd74ea4968ac3bcbfa2eee67aa : FAILURE in 50m 05s","accounts_in_message":[],"_revision_number":1},{"id":"429c026e5ac08cf2801dc198ec5e2b7d4e4d106a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-20 21:17:45.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-20:07:26/720795/1/ : FAILURE in 42m 15s","accounts_in_message":[],"_revision_number":1},{"id":"18096d3e51821737b99c58d702a3e8579441794f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-21 07:33:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4b05e3243d9838cf12e65bc6ff757ae7449e1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:27:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3af96c8f61ad465293b8f79f1a19367c : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/794157154ba7426cab2185d9299ef2db : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e99fde8cec84098a032730408639d79 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68a8e2c161641f2b373c06d59527b24 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6feb611af4ac40a2b56d6b2121de24d4 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/568c810d85864305b0a5d5b54da35027 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c49b41d490445c49c0bfb609e73d38b : SUCCESS in 3m 50s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/e885abb5f24a4eaaa512716a77188313 : FAILURE in 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"dd44609762922ea7aeed10e3c0d1f330edda8f07","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-21 13:38:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"65d12508b9c9b7f3855953def50b178f6b48f1be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 14:35:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9defb077980440eb0f3d43a926d78d1 : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c304146b42b43168a41bafd11c866c1 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56762b3456bd4fc59afa427184622731 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2571bdbee95a46cb9e88b61c204568e6 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd94d2fa3cad4de3b095284dddddc8d1 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9989e4cdc845e5ac7b699e9e00500d : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f976e0e3e24492c9fb6f4246a6bd283 : SUCCESS in 4m 15s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/c32d193b0bf043f286168965a57299a9 : FAILURE in 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fb617cdf91d332625e92cf1b831109913e3ad899","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-21 15:40:32.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-21:02:33/720795/1/ : FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7ffedf741182de44e6e19fd9c3da3daf64677b00","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-27 18:43:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f8bc915038a481b4825fdb9790c3238862f7d513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:42:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cd2b2a62f4a4999b8348e75a9df8700 : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7aa44bcefdc48299419071d20781ca9 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465cb1694cba4256ac4458357d78e763 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e028135bfa244e1baf3a5ad40c68d4e : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/713f25fa65c44673809016fc2f18a768 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8493bc9442dc4b52a6843b47321fec17 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f019193b6b54874b2c35008358ab7c0 : SUCCESS in 3m 53s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/b4f44ed245c24a48b5360ed70c886af9 : SUCCESS in 52m 37s","accounts_in_message":[],"_revision_number":1},{"id":"00a8baf2c9ee1b4fdd96de8e39d512d02085a375","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-28 01:56:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-27:13:42/720795/1/ : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":1},{"id":"456f38118e5deaac7429c784c986d626fa1df55f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-28 14:10:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4819e909a754ad1f51ec799f07f11bc50eb2629a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:10:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e50e1594cba5acd66dacd4de0fda69b31887afe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:04:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adf7ce944cfc4d268740d6c6910a5909 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ef43cade1040669679055be0d4b1ae : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1988394c0314db9886e877367f12bef : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bce8b700079f4e26a1ea38ce1acf2ba1 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c72219bb75c426e8048fe4d569178f5 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/999f5793fb6b4d8c8a8de15eecb00a30 : SUCCESS in 3m 49s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/97b0e65cbe47422f9017518d405e077e : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":1},{"id":"99b54a3c9abe5cf52bab3ace70e018c1e11e7741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:04:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59b37fd0b26328ab977adc430fb40f3837fed752","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:07:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3782b354e9c4482bfba0b43d781798b : SUCCESS in 2m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e504c8a8f064554b18cad1784de8c25 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c5ca0d30550c3a9687ef787480aa1de7fb56e21","revisions":{"4c5ca0d30550c3a9687ef787480aa1de7fb56e21":{"kind":"REWORK","_number":1,"created":"2020-04-17 18:14:21.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/95/720795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/95/720795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/95/720795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/95/720795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/95/720795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/95/720795/1"}}},"commit":{"parents":[{"commit":"a9d499a02aa1dce951e1e8bc30d28951e984593d","subject":"Merge \"Convert sushy tempest to dib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a9d499a02aa1dce951e1e8bc30d28951e984593d"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-17 18:13:59.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-17 18:13:59.000000000","tz":120},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nopenstack-discuss thread\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014237.html\n\nChange-Id: Iace21f058714365115e2cd63c3a463af27540721\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4c5ca0d30550c3a9687ef787480aa1de7fb56e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4c5ca0d30550c3a9687ef787480aa1de7fb56e21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
