)]}'
{"id":"openstack%2Fsushy~721649","triplet_id":"openstack%2Fsushy~master~If4202cc95a2121c4e693aff6777ae19d85f34a82","project":"openstack/sushy","branch":"master","topic":"f8io-ver-up","hashtags":[],"change_id":"If4202cc95a2121c4e693aff6777ae19d85f34a82","subject":"Add import order check","status":"MERGED","created":"2020-04-21 15:08:11.000000000","updated":"2020-04-27 21:00:33.000000000","submitted":"2020-04-27 20:58:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721649-1588021123513-81340386","meta_rev_id":"cfca8af4eb5c44a60629cb6b9c00242747fca769","_number":721649,"virtual_id_number":721649,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-23 22:28:56.000000000","_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"date":"2020-04-27 06:37:31.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-04-27 08:25:52.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 20:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 13:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2020-04-27 19:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":1,"date":"2020-04-27 19:55:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 22:28:56.000000000","updated_by":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"reviewer":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"state":"REVIEWER"},{"updated":"2020-04-24 13:29:31.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-27 08:25:52.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-04-27 19:55:33.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-04-27 20:58:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94f8005872778cef17a5acf15bf9184b3ed69162","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-21 15:08:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b34d3fc6dd7dc7ef2371f5bf3a345a8ff8842cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 15:43:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/208b811881834cae9e1a489c4a414c3e : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/561dba3f9edd41c48ffe2f8299dd86ed : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d79071cff53437cb7dd72738bcad880 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3aa5bf47e94416a73a4e7d56f207da : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/924979f4722344b1b75b3a22cbc27abf : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a58bd0754794781a9802a257e4221e7 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb727f83e4144bf9fdb741a4026e6f9 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bac266a993949b884bc15c9ff3e7197 : SUCCESS in 5m 51s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/e3b12942111648b6afd4ce9fa5e93b15 : FAILURE in 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3e71e0046a73ba17c9c37a66cd5213b6b95a80f1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-21 16:22:37.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-21:10:07/721649/1/ : FAILURE in 1m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d32737d3143665eb81f43d04c17e475cf3f20181","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-21 22:01:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83180c1d1c6de646cd65f53632a343ef07366f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 23:01:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/73243d79521a4a0a89fea9dac18d09bb : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bf4279ea4cd4aabb05d4ab7b2f559e0 : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d5d9ca7c4fd4e1da6951591701e2337 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d40f36ea274e21bc28518dd2280748 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dee70a1b4844e7f9bd1fbd6062cab6a : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbff5c7f4e3144f099c510b30387d874 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef39252ea9d46db83ca03dd75ebe7f5 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68efa847cf5340b6bde1b79d263f0179 : SUCCESS in 4m 04s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/061e21556dc54e219cf4b49f9d0ab372 : FAILURE in 52m 43s","accounts_in_message":[],"_revision_number":1},{"id":"be5e98aca59b20fa69df982d42426120a014a8ab","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-22 04:25:24.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-21:17:00/721649/1/ : FAILURE in 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"fc933721bcc22decafde6a7f9d73e18f64f5f0ab","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-22 08:08:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be7626221bd313eb5a01fcc52ff6f0f933a747d4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-22 08:45:44.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-22:03:07/721649/1/ : FAILURE in 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2ee3b2c99b9a9b173171830113a7b9ae960afe03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 09:02:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f7f6c62505d4b9594f7b1d2dcc3e87e : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18b2ccd58949443dbec7054b05a5bbc1 : SUCCESS in 4m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eff67a42a6714699be38d5f3603a2d6a : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00daef644c64437d80a3bccf358313b9 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3c65587117b43378e89c91503dbe308 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90aef368cebd467180c1ea61ea87478b : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c8f642a808423080c409bcfc7df70d : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/013ec4464f744c4b9dea3ad7b95e5de6 : SUCCESS in 4m 10s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/17d6454f204b48b89e630fd319e50627 : FAILURE in 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"35ac5ac1ec56b135ec3898db17f72703b568471c","author":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"date":"2020-04-23 22:28:56.000000000","message":"Patch Set 1:\n\ndell-recheck","accounts_in_message":[],"_revision_number":1},{"id":"1b6d4df17896b373a8b2c1345462c8ec95541489","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-24 02:00:53.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-23:17:28/721649/1/ : FAILURE in 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"45846656883c05ed408720cbd269502a628c9232","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-24 13:29:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4b80672f60896cf7296750c4ef8d01d78757e8a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-27 06:37:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0b3c50a82267dd6efedde227d4b605296ca3f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 07:44:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5c75d454cf634792aaee532312081a4b : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d4788531945446aa6f0a8be0dd72c5e : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f89ba9051fc4a549b00ed73c1bcbb01 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09c9658c456048cdbfbadca40661ebc3 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4da84f595eb44cdaca1808dd10bf859 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76382d424bae432cbdbb45f2d8d2e4ca : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f72c67b391b4a6cb39e60563163dddc : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae0f5c99f8e84558a3de8f3c5eef2b74 : SUCCESS in 10m 23s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/15dc4ac41e5845bebedb1de5a17a79ee : FAILURE in 59m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5248e230241d1f4ec03f6a4c06634ae22d4317b1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-04-27 08:25:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-04-27:01:37/721649/1/ : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4ad0f3ff36dc80455e314d538ac456831d64a30d","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-27 19:00:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b65cca9281787a75d99c1e22792d1a9bf7245cae","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-27 19:00:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af30dd3b69dd7505915b0a39cdece768bb671c6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:53:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4714f30e4bee4304bb841189eef46c67 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89b14a6e0c24867a56a2b7b7daa3d8c : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/768910e05cce4c6882e0aa3160c45e92 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4fa105f3454e188816ef96c8d758c0 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a004dcd4722f4c66bb60ecbf7fb701e0 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe575e1d53fe47adb796ac9ad3fbc1c2 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d2a4036e514bfc8380965adcc38690 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/382bcc6cbfd6444bb5016ed913b70d85 : SUCCESS in 5m 52s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/d2ee36b28e204aa699f485fa345a1168 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3a64bfc32f5b867fbd2082320129aa3327e46a37","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-27 19:55:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8bcfc578fb7e7220382787058460450cb1c9dbf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:55:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cb21ae8610765b6b349137fc47e4b5cf0987582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:58:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee13bf30f3964f6eab60c6451f0d95e6 : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa074745b251450a9fd76a14da0700f2 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7aeb5121c447b1be76fb1f9b2037bd : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1288b55a3da14a198c483a16d96e05bb : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cedab1a7b9ca491da4946961d384658a : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e94470196fae472786fb3bcf761d51e9 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7dbd02b283e47908c58c015fa374fca : SUCCESS in 6m 20s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/2d978b0c02a14592bf2c457df8645c6a : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":1},{"id":"679655b26345a4c1b4d48bdd554b96a66d4d886f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:58:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9f61c7ecaae5cf06f2854989802d810b4d032eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a4a4e661544bd7979c8559b31c98c4 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5135c5cbc704ce5838792b2454928d9 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bab92f0902480a164d96ed21abcb358f59a94387","revisions":{"bab92f0902480a164d96ed21abcb358f59a94387":{"kind":"REWORK","_number":1,"created":"2020-04-21 15:08:11.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/49/721649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/721649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/721649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/721649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/721649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/721649/1"}}},"commit":{"parents":[{"commit":"010a288b156fa9b3ec11257ca8a2649c4f2305be","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/010a288b156fa9b3ec11257ca8a2649c4f2305be"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-21 13:28:08.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-21 13:28:08.000000000","tz":120},"subject":"Add import order check","message":"Add import order check\n\nSet flake8-import-order version to 0.17.1\nFull py3 compatible version.\nAdd all Python3 modules to stdlib list.\nAlso includes fix to an enum34 dependency bug.\n\nChange-Id: If4202cc95a2121c4e693aff6777ae19d85f34a82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bab92f0902480a164d96ed21abcb358f59a94387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bab92f0902480a164d96ed21abcb358f59a94387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
