)]}'
{"id":"openstack%2Fsushy~727738","triplet_id":"openstack%2Fsushy~master~I7834b49455258f4c3e5690708b23c67d6a361158","project":"openstack/sushy","branch":"master","topic":"oem-loading","hashtags":[],"change_id":"I7834b49455258f4c3e5690708b23c67d6a361158","subject":"Fix OEM extension loading for different servers","status":"MERGED","created":"2020-05-13 11:59:24.000000000","updated":"2020-05-15 16:01:24.000000000","submitted":"2020-05-15 11:35:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727738-1589542551764-fedd4e4a","meta_rev_id":"a4c559fe0729937a989f7f6e9e0fc6d6069623a3","_number":727738,"virtual_id_number":727738,"owner":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-15 09:26:19.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-05-14 14:55:11.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 11:35:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"date":"2020-05-14 10:18:52.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":2,"date":"2020-05-14 08:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-14 08:13:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":1,"date":"2020-05-14 08:59:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 12:02:50.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2020-05-13 14:03:06.000000000","updated_by":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"reviewer":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"state":"REVIEWER"},{"updated":"2020-05-14 08:13:44.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-14 08:59:55.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-05-14 10:18:52.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-05-14 14:55:11.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-05-15 11:35:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f65eef3ed80a7d85eee3611e2c4c7915ce53504","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-13 11:59:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f58b1c0720ddbc9ee170be193fb9a073920ca0e4","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-13 12:01:08.000000000","message":"Patch Set 1: Workflow-1\n\nhave to think about how to unload instances no longer needed","accounts_in_message":[],"_revision_number":1},{"id":"9ccd10e765be4e17da0ee072521eee19bc0757d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 12:58:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6565ee88a4094c5f8a5e548d5d092bad : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ec96d9f6909404ba136d5aa2f589d60 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6d902632db46eabffa22478251684a : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9912558139ce4213bd07a37b794f9382 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff9d9f37d8046af8b06d425b64f5dfb : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0cc9ecd6a354429a95e2216892338c5 : SUCCESS in 8m 09s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/f9b5622401d247c4966c13a2cd904fe0 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9398d1ff671eeba4f8d6dd7c5189eeea48ac0d32","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-13 18:38:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53db8c3e1f561fb28200c758ea015d9d152f1270","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-13 18:41:35.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a803311368366ae8fe5609990c9e0ea3c863c04f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 19:38:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b54afe7112f4bd385066fe369e587ee : SUCCESS in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cef0aea5eb194be3857b563809ecf581 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d629803b7701476399f26bb2b05a82ce : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2fb69984bd648c6ad62d49500b7cba8 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edc57bfa0d5a41faa1e7b9a2a28eb3a5 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209c6683d04e47eba504871d3e25e1fd : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b771952cccf6493c9292a39c4e9d53fb : SUCCESS in 4m 12s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/f6930298f2864eda8e465109b6271e08 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":3},{"id":"22ea5718d76809859ba3a89107fa69d7df54ff48","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-14 08:13:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a4aee4c39f45dbe8aef7345bb63cf4c7d5b0918","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-14 08:59:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"d45cebd3c86354f1f4ceea732ff1fef2ff2abcd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 09:00:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e30c53a744bd70f4bd1db5a8dfded84a9b04442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 10:04:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81fdda2278374390ac5c9c9ebbf13c36 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93219f144b83497b88a11d862b9938bc : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb1817ee4c884d65ac227720a0236cf7 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e16cb49f2b814834b0a78886c2ce73fa : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a282546c70487c9001bebf7acb8f3b : FAILURE in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/745c64c09bcd43449e8dcc8d7b9073bc : SUCCESS in 4m 10s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/17f3190bd0744a5e8531d3e8318ddba6 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":3},{"id":"2ac6a9a22daee38c9be4f78eda78351e19659cef","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-05-14 10:18:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e636806983041090be9fc7f50183b6643a36f3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:18:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebfaa99a1f274368b3cf6fe659adb3b2 : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79c5156a7f1a45699a6c3a22505ddadd : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b7e93cba54435bbed3e88572b609bf : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef50a9f8bb94812b93c661ca2e911f8 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/944ef88136c842f2b5ff0fa11ad0598a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660f3b719fe9416eb47b77ce2bfdd636 : FAILURE in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73c948569f9a438b87a769b97e52974f : SUCCESS in 3m 21s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/4e72c3d57f044da1954a4402bf86cdc8 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":3},{"id":"a741a6133fd0c788140dd19ff09e8f27275e2c8e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-05-14 14:55:11.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-05-13:13:41/727738/3/ : SUCCESS in 50m 50s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-05-13:13:41/727738/3/ : FAILURE in 57m 58s","accounts_in_message":[],"_revision_number":3},{"id":"cdd5f7ac4555b43729c9f948f75f08e3e1feebb5","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-15 09:26:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"db79e8dbb8fd98a3e151ab7205c16a7d77e5d74b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:28:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0aff3b3d2524da4ab7e9a8440e6cdb2 : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f43ad1ec24a4dcd9dd0d3836e72dd35 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd128e9d8c941a78830299ca8348216 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38897e1a349340f297e83443541b2d05 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d77ad84bf2bc42e9aaac47b047063695 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47aa3a1127724aa39df257b94c338798 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b89f2e04f6224780ab6dbce0a9d55785 : SUCCESS in 3m 41s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/ffd63cf1826a44c5b7e1d0b7733615ef : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":3},{"id":"e4e02c30faffe4d36cce0913e6c72298aa5e4785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:28:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d56905b668a04a1789daf8be349d861c909a6fd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:35:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ca0f86d2ae543b081736d0c3e3e8f4a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d263fb1f0c44350baf3dcefc9e6d70d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5efc91b559814a21aca5341a5934cb39 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b43cd1405b49c7897be0795f3302e8 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a022d413324e1c8e94d3408ed36f6f : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66c08edaddcb40ac96dc143fcd2d42b4 : SUCCESS in 5m 30s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/2bd3632503ec45269a9d5cef154f356a : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f4046072ead60aadaa2c20363c14c6ea00fbfd63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:35:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8e0b8948794c90f6eb2c856d63ba86461e2a1ab2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:37:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f756e95c949e4fb7966f91dfd6afd048 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/109cc6e58fe94105a8cdca3e7f638663 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3},{"id":"52dcb4dff37c021f0e2515288660a0ef811cf488","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-15 16:00:52.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 40864e1b91ad57e310f260ae505e46c9231f67f5","accounts_in_message":[],"_revision_number":3},{"id":"f40203f9945fcfc9b2fbed71b0be43dd10c6b652","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-15 16:01:24.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 0241cd9a43d4935250c92c29986faa0287923f0d","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"359f2ec2c11cf4e4d2677c2af4ed715e92ce33f3","revisions":{"56feda70e75f330b40f401362b5f543b00623cf0":{"kind":"REWORK","_number":1,"created":"2020-05-13 11:59:24.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/38/727738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/38/727738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/38/727738/1"}}},"commit":{"parents":[{"commit":"98e907ec5abf1aeeeea3a0231787e1bc61b75cf5","subject":"Fix pep8 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98e907ec5abf1aeeeea3a0231787e1bc61b75cf5"}]}],"author":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 11:52:09.000000000","tz":180},"committer":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 11:59:22.000000000","tz":180},"subject":"[WIP] Fix OEM extension loading for different servers","message":"[WIP] Fix OEM extension loading for different servers\n\nInstead of keeping 1 instance per resource type\u0026vendor,\nin extension object store instances also per parent resource\nso that when using sushy with different servers their loaded\nextension points to the right server not the 1st one loaded.\n\nChange-Id: I7834b49455258f4c3e5690708b23c67d6a361158\nStory: 1689605\nTask: 39761\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/56feda70e75f330b40f401362b5f543b00623cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/56feda70e75f330b40f401362b5f543b00623cf0"}]},"branch":"refs/heads/master"},"c3f0f58b420b7c573bc38372bae1f0a3a2e1b044":{"kind":"REWORK","_number":2,"created":"2020-05-13 18:38:33.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/38/727738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/38/727738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/38/727738/2"}}},"commit":{"parents":[{"commit":"98e907ec5abf1aeeeea3a0231787e1bc61b75cf5","subject":"Fix pep8 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98e907ec5abf1aeeeea3a0231787e1bc61b75cf5"}]}],"author":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 11:52:09.000000000","tz":180},"committer":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 18:36:15.000000000","tz":180},"subject":"Fix OEM extension loading for different servers","message":"Fix OEM extension loading for different servers\n\nRemove caching and re-using loaded extension objects\nas it loaded the same instance for different servers\nmaking it impossible to use it for more than 1\nserver.\n\nThe caching of extension manager is still in place,\nconsumer of loaded extension manager can manage and\nreuse the instance as needed.\n\nChange-Id: I7834b49455258f4c3e5690708b23c67d6a361158\nStory: 2007669\nTask: 39767\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c3f0f58b420b7c573bc38372bae1f0a3a2e1b044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c3f0f58b420b7c573bc38372bae1f0a3a2e1b044"}]},"branch":"refs/heads/master"},"359f2ec2c11cf4e4d2677c2af4ed715e92ce33f3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-13 18:41:35.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/38/727738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/38/727738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/38/727738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/38/727738/3"}}},"commit":{"parents":[{"commit":"98e907ec5abf1aeeeea3a0231787e1bc61b75cf5","subject":"Fix pep8 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98e907ec5abf1aeeeea3a0231787e1bc61b75cf5"}]}],"author":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 11:52:09.000000000","tz":180},"committer":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 18:40:29.000000000","tz":180},"subject":"Fix OEM extension loading for different servers","message":"Fix OEM extension loading for different servers\n\nRemove caching and re-using loaded extension objects\nas it loaded the same instance for different servers\nmaking it impossible to use it for more than 1\nserver.\n\nThe caching of extension managers is still in place,\nconsumer of loaded extension object can manage and\nreuse the instance as needed.\n\nChange-Id: I7834b49455258f4c3e5690708b23c67d6a361158\nStory: 2007669\nTask: 39767\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/359f2ec2c11cf4e4d2677c2af4ed715e92ce33f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/359f2ec2c11cf4e4d2677c2af4ed715e92ce33f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
