)]}'
{"id":"openstack%2Fsushy~728486","triplet_id":"openstack%2Fsushy~stable%2Fussuri~I7834b49455258f4c3e5690708b23c67d6a361158","project":"openstack/sushy","branch":"stable/ussuri","topic":"oem-loading-stable/ussuri","hashtags":[],"change_id":"I7834b49455258f4c3e5690708b23c67d6a361158","subject":"Fix OEM extension loading for different servers","status":"MERGED","created":"2020-05-15 16:00:52.000000000","updated":"2020-05-16 16:36:33.000000000","submitted":"2020-05-16 12:55:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728486-1589633701624-f8705a57","meta_rev_id":"09e6e81bae4e9a46e59ef2ba9780b6301affcdaa","_number":728486,"virtual_id_number":728486,"owner":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-05-16 16:36:33.000000000","post_submit":true,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-16 12:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-05-16 12:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-16 09:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-05-16 12:09:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-16 09:45:59.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-05-16 12:09:59.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-05-16 12:55:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-16 16:36:33.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"}],"messages":[{"id":"bf3b1b4bb859debabc52480aad1106708c0df048","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-05-15 16:00:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3a0b825e3de79c80ff059db62bbd4dcdbc3d45dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 17:04:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fab892a391e4c09b16f14fbb049ad50 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2638981fc1b04475ada75d6fdde42cfe : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa899ea207449848e1e0c2cd34b99fa : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcf90826c1244eee9d7e0f911873aeaa : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6701ca5a6c6423189d113e15130c75b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19532c5426f64f6eaff88d8acd30c8d2 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d57ee53eb7764ac5b8e972ab82a1f278 : SUCCESS in 5m 13s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0758c17a6c9042e9bd8017245633a725 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d0e8ca7e53365277774621cf9509cf191302fe68","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-16 09:45:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"587f70485cd69621eccaf8a8d543376064a972d4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-16 12:09:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ffdf9e990eb3b1382966f48f8cf6fadf11afa62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 12:10:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88a4a55a60bf6800bde2054ab15cad65db709d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 12:55:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e0e8f4bef476e4d6b7dbdcc53d8ae24cf73d301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 12:55:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f1c221a1f834c7ba058e205201ee2bf : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c651ae9352489dbffe790b1a285045 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70200b70b7354afc8f221900fa4b01a9 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd37b88df52740afa92728c0d549cc2e : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b25249d1ce5b41aca37e38da8d284a55 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5d355b87d7e4199a009e9f8fcf32e55 : SUCCESS in 3m 30s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/3f1d9887f17c42febc5c57c11787e486 : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f33ac013757576776d878054593eec9dba21640f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 12:56:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f69c4ff8314d29b8068edaf9dfef96 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8d65d71e819436d9716ad16090b3064 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a52af6a9bf0e75def3efe04ac58ae6093da1225a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-05-16 16:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-05-15:11:00/728486/1/ : SUCCESS in 51m 33s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-05-15:11:00/728486/1/ : SUCCESS in 52m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40864e1b91ad57e310f260ae505e46c9231f67f5","revisions":{"40864e1b91ad57e310f260ae505e46c9231f67f5":{"kind":"REWORK","_number":1,"created":"2020-05-15 16:00:52.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/86/728486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/86/728486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/86/728486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/86/728486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/86/728486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/86/728486/1"}}},"commit":{"parents":[{"commit":"90b72f995c022f40ea06eca898eabc97a8c7b5f9","subject":"Merge \"Fix pep8 test\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/90b72f995c022f40ea06eca898eabc97a8c7b5f9"}]}],"author":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-13 11:52:09.000000000","tz":180},"committer":{"name":"Aija Jaunteva","email":"aija.jaunteva@dell.com","date":"2020-05-15 16:00:52.000000000","tz":0},"subject":"Fix OEM extension loading for different servers","message":"Fix OEM extension loading for different servers\n\nRemove caching and re-using loaded extension objects\nas it loaded the same instance for different servers\nmaking it impossible to use it for more than 1\nserver.\n\nThe caching of extension managers is still in place,\nconsumer of loaded extension object can manage and\nreuse the instance as needed.\n\nChange-Id: I7834b49455258f4c3e5690708b23c67d6a361158\nStory: 2007669\nTask: 39767\n(cherry picked from commit 359f2ec2c11cf4e4d2677c2af4ed715e92ce33f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/40864e1b91ad57e310f260ae505e46c9231f67f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/40864e1b91ad57e310f260ae505e46c9231f67f5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
