)]}'
{"id":"openstack%2Fsushy~730990","triplet_id":"openstack%2Fsushy~master~Ib525de0e7638ca3affcd08512be1159cebacd38f","project":"openstack/sushy","branch":"master","topic":"task/39863","hashtags":[],"change_id":"Ib525de0e7638ca3affcd08512be1159cebacd38f","subject":"Make Volume/VolumeCollection operations blocking","status":"MERGED","created":"2020-05-26 20:53:36.000000000","updated":"2020-06-10 09:45:07.000000000","submitted":"2020-06-10 09:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730990-1591782178817-db77c85b","meta_rev_id":"3efb6145eeb7cbfa9f8cbda04fe4ca21c99f255e","_number":730990,"virtual_id_number":730990,"owner":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"date":"2020-06-08 21:58:03.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 09:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2020-06-10 02:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":2,"date":"2020-06-10 08:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":1,"date":"2020-06-10 08:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 21:58:03.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-06-10 02:05:36.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-06-10 08:33:31.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-06-10 09:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a16ddd474cba246c57015c9b591545d008c3d52","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-05-26 20:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"616f458c4100160f93745fc6d93df2de7e3fc7a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 22:32:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d61106b5e9724d479ce50eaf6d236c1f : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db82920060f444d99b8ec8206b6f0024 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f3b53b32f245a897999bc8d29caff3 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fda00cd721fb40a5a992e64275c1d104 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2790e01b321548e1933d274ebd393048 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8489c3d79814422ab179f495a7033cc5 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17b1e0ef25794043b2013a3fe1f8a787 : SUCCESS in 5m 08s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0ef4ba06889b4653b102dc991895313f : FAILURE in 1h 05m 53s","accounts_in_message":[],"_revision_number":1},{"id":"5604bf4d78a6b32acf86aa674f809d97201d78f6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-06-08 13:45:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"87a47a5dad9e794de4abb1b51d05d257c62991c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 14:56:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a06a015b371a44c6851fb9bfe21d1729 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59271abe3634c4784902e5e48556bb1 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c054fa27414b4e4caf209fa6595fc1aa : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d837017cf5ca4c9b8030a55358e28084 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c9a45530c284e069357fd957877dd36 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36512df004b64334a495152018ba5b1d : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd0cd67aa34d4299a01b5675f304efb6 : SUCCESS in 5m 22s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/e75b8f285c174b489826ef0bcf308af7 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bd1460f1d939d8ff369e8a367ce336902b66312c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-06-08 21:58:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-06-08:08:45/730990/1/ : SUCCESS in 48m 20s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-06-08:08:45/730990/1/ : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":1},{"id":"75c4afbc11e2b6aff85690b5d04c846a580a3f02","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-06-10 02:05:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7df3ff8b4617fa608dc196b828753e394c7ae83c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-10 08:33:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02ec90510c11e883d4cc2beacdf93a51867a390f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 08:33:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4709963b3c214c44d31044336fcfca2b19b2e2ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 09:42:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec5143f6f99b41e8b978f6efc8c26933 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0d197aa5d74e009a905a778c083a53 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8845d048536e41d8bb311d3c0cbd491f : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb31221e84f473d9efb037cb4a7c62a : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66173253ca4d4360bae58292238fe8df : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d03c63b558e467a82914f047c7a4044 : SUCCESS in 6m 00s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/26064a642820440f9607f5f5328d96c1 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7dc560e2e91dfd1eaa0304a18f713aeb289a0fcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 09:42:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"67e1377345f0b1056d9648b7ec1d5c8e1882583f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 09:45:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7f1af95af1469f8cb823d2813d9fb1 : SUCCESS in 1m 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6edc7f78d5f42ceaadb83f2c2259d2c : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"461a962be9a396bf96f36238108dcddd0fb2aa67","revisions":{"461a962be9a396bf96f36238108dcddd0fb2aa67":{"kind":"REWORK","_number":1,"created":"2020-05-26 20:53:36.000000000","uploader":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"ref":"refs/changes/90/730990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/90/730990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/90/730990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/90/730990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/90/730990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/90/730990/1"}}},"commit":{"parents":[{"commit":"4fa057b57f3aa457bce804454c89f17932e2e047","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4fa057b57f3aa457bce804454c89f17932e2e047"}]}],"author":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2020-05-26 20:49:06.000000000","tz":-300},"committer":{"name":"Bill Dodd","email":"billdodd@gmail.com","date":"2020-05-26 20:49:06.000000000","tz":-300},"subject":"Make Volume/VolumeCollection operations blocking","message":"Make Volume/VolumeCollection operations blocking\n\nThe operations in Volume and VolumeCollection (initialize_volume,\ndelete_volume, and create_volume) may be handled asynchronously by\nthe Redfish service. This patch will leverage the new blocking\nparameter in the sushy connector module to make these POST/DELETE\ncalls blocking. The logic to do the task monitoring will then be\nhandled in connector.py.\n\nChange-Id: Ib525de0e7638ca3affcd08512be1159cebacd38f\nStory: 2003514\nTask: 39863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/461a962be9a396bf96f36238108dcddd0fb2aa67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/461a962be9a396bf96f36238108dcddd0fb2aa67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
