)]}'
{"id":"openstack%2Fsushy~745950","triplet_id":"openstack%2Fsushy~master~I485d56a9804af723ddb55f8bc26f28a5ebefccc7","project":"openstack/sushy","branch":"master","topic":"fix_firmware_update","hashtags":[],"change_id":"I485d56a9804af723ddb55f8bc26f28a5ebefccc7","subject":"Allow monitoring progress of a firmware update","status":"MERGED","created":"2020-08-12 19:12:34.000000000","updated":"2020-09-04 23:30:09.000000000","submitted":"2020-09-04 23:28:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745950-1599262113035-038e877c","meta_rev_id":"dc73d3cf91903150f00d08eb9996d2e5aebd33f8","_number":745950,"virtual_id_number":745950,"owner":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"date":"2020-09-04 23:02:12.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 23:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"all":[{"value":1,"date":"2020-09-04 21:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2020-09-04 20:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},{"value":1,"date":"2020-09-04 20:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2020-09-04 20:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":1,"date":"2020-09-04 22:27:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 19:15:04.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2020-08-12 19:19:24.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-12 19:19:45.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-08-12 22:09:53.000000000","updated_by":{"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},"reviewer":{"_account_id":28533,"name":"Mike Raineri","email":"mraineri@gmail.com","username":"mraineri"},"state":"REVIEWER"},{"updated":"2020-08-12 22:21:44.000000000","updated_by":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"reviewer":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"state":"REVIEWER"},{"updated":"2020-09-03 18:51:35.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2020-09-04 20:46:22.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-09-04 20:52:54.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-09-04 21:30:30.000000000","updated_by":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"reviewer":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"state":"REVIEWER"},{"updated":"2020-09-04 22:27:36.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-09-04 23:02:12.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-09-04 23:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ec89b99428802684822f017588e271d471086a3","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-12 19:12:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6614d17c1b732fa38adde2fe97f38792bf234801","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-12 19:21:37.000000000","message":"Patch Set 1:\n\nHey folks, could you take a look at this WIP patch if you have a minute and see if you think this is the right direction to go? I\u0027m working on updating the test cases, but that work turned out to be more than I thought, so early feedback would be appreciated.","accounts_in_message":[],"_revision_number":1},{"id":"f5cba325f2df7f5f9bca1426b424cbe4e891f77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 20:44:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af26f4f22af3405282c67b2506fcd1c9 : FAILURE in 3m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0af9c1a3fc3e4143af0bcf9e3b91fac6 : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff5ce1e3709493595a5843ac56bd5e4 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f95c5306317f4feaa637eedaf97fe7f1 : FAILURE in 16m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8202ec5cc3c46a984a1f5e62ee3051e : FAILURE in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb0aa434e42c45578e18ac96dd344f77 : SUCCESS in 11m 29s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/b5cefdea5cee44aa9b967dccedf8678f : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3e6684e855ac197f4763cf6d3ae8a0171cf18014","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-12 22:43:04.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-12:14:12/745950/1/ : SUCCESS in 58m 41s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-12:14:12/745950/1/ : FAILURE in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"93aeccc3f98ac187fd6436917c84d23e2a826911","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-13 22:11:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"173fb421d9c5a580068cb4aa6c43f049ce2c1e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 23:18:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/093ae89c014f4db9af09d628dfd5c66d : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbfd7280455c45fc9dd0502c45b9976b : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c871d1ee2f4d528d7e3592809c7b66 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b57fd52cc8b2491caf92954eedcc397c : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15091530817f47f0a4fbe68ff61c3879 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0cf2e03f25c41428a107b16a12ff84b : SUCCESS in 8m 40s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/9cdcc7e8b9c24e0aaab33168eaebf3ca : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":2},{"id":"75ccdb57973801aee43eb0c8ddef8fc1bceed70e","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-08-14 01:28:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"adfd4bd965a5963810975696df02488956e8b41e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-14 01:38:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-13:17:11/745950/2/ : SUCCESS in 50m 55s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-13:17:11/745950/2/ : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":2},{"id":"492f0e518a93cadc9d65ca0c528b405397459305","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-14 21:36:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6684580f179ce7d4fed9a038316f69f5907eac0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 21:58:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a347c7aba2644839d119d65c64c614b : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/806178bceae5465abaa9b68386beeb09 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d5567fe34243b6bf55683357d861ce : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1966402c71714899ad9b46d818fe1d5f : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf2c24460622457d876f31f9d9287d5a : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4b3807bd0d4f7284802f3028ca5515 : SUCCESS in 6m 00s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/fa90eefc862a4bc49cdfe3c4d362a64b : FAILURE in 16m 23s","accounts_in_message":[],"_revision_number":3},{"id":"32de6fab263fb0382846da9713e043b801553d47","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-14 22:30:37.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-14:16:36/745950/3/ : SUCCESS in 54m 03s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-14:16:36/745950/3/ : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":3},{"id":"14aad12502e910fcfc060316713f904e917a6bd4","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 13:15:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cb428a308530309b3692454d94bb56ce36f0443d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-30 13:16:41.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-30:08:15/745950/4/ : FAILURE in 1m 04s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-30:08:15/745950/4/ : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":4},{"id":"053bdc3f6ae67eb9923696b51d6b8673fbd04a58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 14:02:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afb365bd54674d11a7dedef843d40a04 : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d48dda7f90ca459593b473179c0569e3 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08ffc6c383c47979d2ccb5f3851a50d : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50c072bedc29493c8cd20ebd06a6b51d : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0516b0f6f30642e5a95429902e45258c : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8267804577140c9966ab93229bbc82e : SUCCESS in 5m 56s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b8e80fbfad00446584f7174bd20b703c : SUCCESS in 45m 16s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a187ab6443cc4605a8c84378920c4c12 : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":4},{"id":"349a5331fdbc12e90ba5e3fa8ce48b63fe79fc66","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 16:11:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"28348692a0a3e96dc10dd6b4204f688bd2607ab9","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 16:16:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3a60923f43fe6fbbf23689ccc8e33fa03322b5c4","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 16:20:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3d14ecf617bf53c6acda46997f4655067257ddb1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-30 17:18:55.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-30:11:19/745950/7/ : SUCCESS in 47m 47s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-30:11:19/745950/7/ : FAILURE in 58m 42s","accounts_in_message":[],"_revision_number":7},{"id":"25699c834c04a97ac2223cb2c2c39ef743864e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 17:19:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d03bd6494fa466ea4755624d4bc9609 : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd447d31b44a44bab18422c52be56da4 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e71e31d97824f999068729e8ec0d449 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57afa0c8d9f146cd8bd29432ca7252f9 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4081707440f84454a4d4a85fe2a2ddff : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d59860ef80045af9a7f7558e63e097c : SUCCESS in 7m 17s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/66c060abc5f74cc2820c09126da596db : SUCCESS in 37m 01s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c3af2324447b49549fb61fd8b8c7fb6b : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":7},{"id":"138e841235bbf826c0e55e148d97e93d046cf7d1","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 19:32:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d1189a168f802f6c11e90c9f98c703fb55424050","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-30 20:29:29.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-30:14:32/745950/8/ : SUCCESS in 51m 04s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-30:14:32/745950/8/ : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":8},{"id":"8c5ff9a08a94defabe776eb340d23a0e89c443ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 20:31:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caada586ff984006b621801052c24198 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d394b91e9c345e5811d33dd670b1561 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a7e67a24d1458290a3fc6132cff269 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11d31cf748c446348548079e323e3650 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18f3c9a91a9042a4937f26c20e316b78 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3187a993a59746309df11c00e72e24ea : SUCCESS in 6m 27s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d8e787c3959043218cb5690c90f676ff : SUCCESS in 56m 08s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4e8921b559c84a188b311f33e8f7dc4e : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":8},{"id":"1ad790e21f2cdb73ccd5e602a1178f7d0e934412","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-01 13:57:19.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"d4406b78e3278cf2d4e1434d55a45f6fe7c0fa45","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-01 14:14:16.000000000","message":"Patch Set 9:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-01:08:57/745950/9/ : FAILURE in 16m 43s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-01:08:57/745950/9/ : FAILURE in 16m 45s","accounts_in_message":[],"_revision_number":9},{"id":"e7ab340d693c270344956567f70307285f9c91d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:05:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/172ecc1e7500420db2a4b0dae7f0967a : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/636d10e532b44b7e89c5ed926351a176 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07fd53c5ab22407fb2ed1cfca4622625 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3794f9417d5740a99e5599db114e0b07 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ae2bd78fba4584a76a829eaf69622e : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0898ace1e2fe499db0598038a2f0a1bf : SUCCESS in 6m 53s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d4ff5b3b0eaf4be99ba5f989a18d6058 : SUCCESS in 49m 24s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/57c72a4f2ba84f3c9f08c90ebb36d0e9 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":9},{"id":"e838e962eaad2b8b919ede0f4f5586d95457b9fe","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-01 22:20:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"74d85c81b549a0ebf1d998de9df37f3e049a2844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 23:18:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/760773ac3e5f44baa9ed0eee2b8ba405 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/554406bbc9904d63936b5ff9f7c0e59c : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961620a4f1b64bdfa1e7c7826fc39f53 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0aea1a792cc9457397223b6490b0983c : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da9223d4b904e6489989c15d7605810 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355de05782d74bd69a0dde99300fac0c : SUCCESS in 6m 53s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a8944129658d4a2a8635162dfca49000 : SUCCESS in 51m 52s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/57bcd71e017b4323bcebdffc53ab8123 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":10},{"id":"93ba8648254345694363f5daf477b343aca025ad","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-01 23:22:33.000000000","message":"Patch Set 10:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-01:17:20/745950/10/ : SUCCESS in 59m 19s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-01:17:20/745950/10/ : FAILURE in 1h 02m 11s","accounts_in_message":[],"_revision_number":10},{"id":"7567eb18c8ffd6bc79ecdb275c5b36a24e6b0501","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-02 01:22:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"63df5063b2a92939b048882c99343c722563129d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 02:16:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/786904c7fc054285a8823a903275fffe : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a233406ed3bd4c618249214fbe57aa87 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/737e126e60ed46768add3e6a3dd40f03 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b8de0553b44cc4bba182e26acfd3e5 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f4c8a53251f4763af1157242e00fd3f : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02860df82ed84581857a50aacd5e7f21 : SUCCESS in 4m 54s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1bdd3f00d9e4403a86c6fe362b764d74 : SUCCESS in 53m 07s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/99a8a9ffc3ab44ed8bac0a6b0dfb8d9b : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":11},{"id":"a3e4d7289e0209e946ce752c7f8b36466e765367","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-02 02:30:27.000000000","message":"Patch Set 11:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-01:20:22/745950/11/ : SUCCESS in 1h 00m 41s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-01:20:22/745950/11/ : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":11},{"id":"c11626614b9eb693044b6f06dc4a6bc7acd55b01","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-02 18:12:20.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b565f14a42bada7379154a09581fdd50fae82a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 19:18:23.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/828455d8451a4083bef97fecb0c755d4 : SUCCESS in 6m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4f2e316808e40ce912e6a04a561b4a5 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da8c5096a7e41aea60edbff231f841b : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a485a98f6974388ae18068faebdaa41 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79b4572b35d64552953a55ea93f3a76e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587fc66add104b04bd447149a41a8a9a : SUCCESS in 6m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4ae68ffdf40541b592d7237b8b8b23f9 : SUCCESS in 55m 37s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/30f66f155c104d5c9fb71651add394ed : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":12},{"id":"3ada59549e5aea9b70862ab0a7050dc08b1301a8","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-03 02:04:33.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"9a6ad3604228843d3850aa672dc6b39c3044d580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 02:46:22.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6f56d929e8c42dd8604daa0b57c490b : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d2071ecdc8141f49e42886f530be00d : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a847d8625982427494624aa6149cec30 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fb570d6056c43c18038df820e981be7 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f3ad3034ec2478aa4609985c48a8401 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f0d74747304654b5a6178027b7c8cc : SUCCESS in 6m 01s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f8a3c0e619f64952bb406f7b7a5b52fd : SUCCESS in 37m 23s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7cf4db553739445ab7f91ff4120ceb5c : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":13},{"id":"22b267b2b31779bfb60258925d9f82b378788275","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-03 17:14:49.000000000","message":"Patch Set 13:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-02:21:04/745950/13/ : FAILURE in 1h 11m 07s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-02:21:04/745950/13/ : FAILURE in 1h 29m 04s","accounts_in_message":[],"_revision_number":13},{"id":"f5010629509c1ff434be98c77c7ed28fd39a167b","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-09-03 18:51:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c94af71c481bc28abdc8039a300575ce5d521939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 20:03:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51538150d2994be28d67292810ef1098 : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37082066197d4b49abec2c4358053c67 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bca4fcfb9da4e7ca16a557d07bd0640 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/840fcccf15aa4df8a22d427ae18f5a59 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbeabfefdc0a4b4da0c11231a807d9e9 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c85eb748c1482a969518772c13d8d3 : SUCCESS in 6m 25s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a23174f177d04db18a9200ae08cfd171 : SUCCESS in 48m 48s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/054311c52e324e91b405f5ed82a375c2 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":14},{"id":"68b99c9db504f8dd05d6435627f2bccdd316d456","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-09-03 20:50:12.000000000","message":"Patch Set 14:\n\n(7 comments)\n\nA few comments...","accounts_in_message":[],"_revision_number":14},{"id":"e818325149a351da16f9bda1a3fdc17833f1379b","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 02:16:04.000000000","message":"Patch Set 14:\n\n(7 comments)\n\nThanks for the review Bill!","accounts_in_message":[],"_revision_number":14},{"id":"352f8cfc9c27f7d5eb8fba17f2ce8ff4a4a501d2","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 04:03:59.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3eb544029ef1515eeabf16f9a48d0d9788d1a7ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 05:18:41.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f91ca424ad1e47fd82140bbd79e6f20a : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56669892457f4fde94874555775a061a : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43e5b1395074e3babdf67b10e044a1d : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db8647658636477bad2518a4767b571d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf172cd8a379457c8489b1ebe12005f7 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af5afc4e5df4eefb5b2c5aaa6241a1a : SUCCESS in 7m 18s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6f834020fe9a45b2992bb16eeec1bf29 : SUCCESS in 1h 08m 46s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/47a053b9014441cab2528687ea500a8c : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":15},{"id":"e046c61a924de050d3cadf0f74d36950d7801182","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-04 10:37:28.000000000","message":"Patch Set 15:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-03:23:03/745950/15/ : FAILURE in 1h 07m 53s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-03:23:03/745950/15/ : FAILURE in 1h 40m 50s","accounts_in_message":[],"_revision_number":15},{"id":"1c554fc556a23aedf7e880c90dd7659bb27437c8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-04 14:43:19.000000000","message":"Patch Set 15: Code-Review+2\n\n(2 comments)\n\nLooks good. I\u0027m happy with this minus the fact it is missing a release note.","accounts_in_message":[],"_revision_number":15},{"id":"c1bd9ca41ac34410288e6da5c2ed458e15d553c1","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 15:16:31.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f374a89cecb936b6377aa4634bb016c2d6b67c14","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 15:20:13.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"956bfcc7c4c3e097ce39990ad88a2073a3f5501a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-04 16:04:15.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"ac0a98e0ef2073b39aa6d8eaea05950d3e6b4b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:19:18.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5751fcfa1a314de4b1f9a3ba8813d8cb : FAILURE in 4m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a79f888a10e4040909bbc16986a5d95 : FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9bb6cf504649faa43d30aca7c51fda : FAILURE in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a60bf191eca74e989402de4ca34ed6a0 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9ab44da92f4e3f8c1572ed270d43ce : FAILURE in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167765f4fe204336970314337d277ff2 : FAILURE in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c137bd91b8a41a383dc547250bd07db : SUCCESS in 5m 26s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/43df584b245242f59c2f828d37040416 : FAILURE in 53m 16s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/32f23938555c4cda99a15b7d85c6468a : FAILURE in 55m 47s","accounts_in_message":[],"_revision_number":17},{"id":"00adcd451ba4e4ad224ad3fab03204e44733122e","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 18:37:22.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"fdead5960acee3f6927485a6ae1a622e2bda55c1","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-09-04 19:04:18.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)\n\nLooks good. The one comment I added about the Allow header could be fixed in follow-up.","accounts_in_message":[],"_revision_number":17},{"id":"4ed811cf05170bcd4e134167f770d7e844e3945b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 19:08:24.000000000","message":"Patch Set 17:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ab10c51714473c960f048d3711c926 : FAILURE in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b175d4d131c4c2d9061e791d1e316d2 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/856e7fdc73484dd98c6bca07d1d933e0 : FAILURE in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567468bc90a9468da8c09f10e099d469 : FAILURE in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dffc875e653f4be9b7361f39c808a325 : FAILURE in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916b0492789248359bbf7c20e15991f5 : FAILURE in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dca449cde1840bcab459a75410bb337 : SUCCESS in 4m 51s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/aac61601775a4b21901bc1804a907853 : FAILURE in 26m 42s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/58a40cbfb5914d0184118a62aab97b98 : FAILURE in 28m 54s","accounts_in_message":[],"_revision_number":17},{"id":"dce88e25d159eaf3c26a7c61fa6320b0e041929f","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 19:33:11.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"41d8b3559f455857768994cd12f5d4df0bc45b93","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-04 20:04:07.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"57f6441e797692ebdfb1bfa5176211aafb87a0a8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-04 20:46:22.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"01a750b01712518b7ab7792bfe17f56d9d85c0a2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-04 20:48:29.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"bb8e8a5444bcd30b700237726b2a85d6edea3ff3","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-09-04 20:52:54.000000000","message":"Patch Set 19: Code-Review+1\n\nNice job Chris and Aija.","accounts_in_message":[],"_revision_number":19},{"id":"fcf929fc1402b76a25d44fbcc35ae18e34ac4775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 21:05:39.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/defd5c9a3dcd411fb0b7d71f7a862f4e : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9177a3ab165f4faeb0e2956d006e2b8f : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b530ff9ce4f4636b4ff43a72cfd0f0f : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49dd28053b524677a20ca40455a43960 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b5c3460ff954034864e3f0518e7885b : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa19427e18a34339ac2c4bb006e00b95 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76a8574f3b34dafa1acfe58530df11a : SUCCESS in 3m 29s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5363bfaf7b3a46438ec684254b93e7b5 : SUCCESS in 53m 01s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b26f079d27db45f8bf0625a474e46c52 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":19},{"id":"6b5bd114deca0d7265ab7da8a75794f4d3850adb","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-09-04 21:30:30.000000000","message":"Patch Set 19: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":19},{"id":"ea8bc98220cce04962bba2a6531c82aa8caff872","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-04 22:27:36.000000000","message":"Patch Set 19: Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"0558245f1f8c4e7f1cc62b124508b497b01e5bd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 22:27:47.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"50057237430f06342c6bd4273a1c39b74b367333","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-04 23:02:12.000000000","message":"Patch Set 19:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-04:15:03/745950/19/ : FAILURE in 19m 24s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-04:15:03/745950/19/ : FAILURE in 15m 35s","accounts_in_message":[],"_revision_number":19},{"id":"689a72bf389b96b54afdcd91f0d1fc08b9ab772f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:28:32.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/796b9e345d01466494f6ccfb00e867fb : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ce3770150440328bc03b4e67fa098f : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4728fd5c1cfa4bdd9a501b4ec31841c9 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb2db3de116d4190a2349fe39c88a242 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a81717fe47a41709525fddc11ca0a72 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d1a3f5401b14eb7814e239e604f9cdf : SUCCESS in 3m 33s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2ea44e35b98f4dbeb137e44949477952 : SUCCESS in 48m 08s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2db21119685c4f11b75472f24172fb42 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":19},{"id":"9421b71dce0bf9edd3ba165bfbca52e4d37810c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:28:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"7d73f483d532f7761cfa83be52e735ea244e2514","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:30:09.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2fa0fbe2bd473aa798aea4550f7c98 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2746715c357643bf9daa518804c4aa07 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"3f052a34982628d5dd2ffe5c4da742e5aa65dc75","revisions":{"bdee090f028e5d88d04b00d515de1579b2e41ba1":{"kind":"REWORK","_number":1,"created":"2020-08-12 19:12:34.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/1"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 19:12:26.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a Task object so that the update job can be monitored for\ncompleteness. A new Task object is added according to the Redfish Spec,\nand StatusCodeField and HeaderField classes are added to provide the\nability to populate fields from the HTTP status code and headers\nrespectively.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bdee090f028e5d88d04b00d515de1579b2e41ba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bdee090f028e5d88d04b00d515de1579b2e41ba1"}]},"branch":"refs/heads/master"},"1426b1c1d5c80c37b62a52faa4ab2867d55fac40":{"kind":"REWORK","_number":2,"created":"2020-08-13 22:11:09.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/2"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-13 22:11:07.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a Task object so that the update job can be monitored for\ncompleteness. A new Task object is added according to the Redfish Spec,\nand StatusCodeField and HeaderField classes are added to provide the\nability to populate fields from the HTTP status code and headers\nrespectively.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1426b1c1d5c80c37b62a52faa4ab2867d55fac40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1426b1c1d5c80c37b62a52faa4ab2867d55fac40"}]},"branch":"refs/heads/master"},"883a7196eda54391f66064b9976a267d6463d0d5":{"kind":"REWORK","_number":3,"created":"2020-08-14 21:36:18.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/3"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-14 21:35:23.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a Task object so that the update job can be monitored for\ncompleteness. A new Task object is added according to the Redfish Spec,\nand StatusCodeField and HeaderField classes are added to provide the\nability to populate fields from the HTTP status code and headers\nrespectively.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/883a7196eda54391f66064b9976a267d6463d0d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/883a7196eda54391f66064b9976a267d6463d0d5"}]},"branch":"refs/heads/master"},"6b9b00b66ffd5ba27f9e3e4f6a2551460780c5c1":{"kind":"REWORK","_number":4,"created":"2020-08-30 13:15:24.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/4"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 13:14:33.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a TaskMonitor object so that the update job can be monitored for\ncompleteness. New TaskMonitor and Task objects are added according to\nthe Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6b9b00b66ffd5ba27f9e3e4f6a2551460780c5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6b9b00b66ffd5ba27f9e3e4f6a2551460780c5c1"}]},"branch":"refs/heads/master"},"57deb16e7d3c0ee611d4d267662217e41cb55954":{"kind":"REWORK","_number":5,"created":"2020-08-30 16:11:54.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/5"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 16:11:49.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a TaskMonitor object so that the update job can be monitored for\ncompleteness. New TaskMonitor and Task objects are added according to\nthe Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/57deb16e7d3c0ee611d4d267662217e41cb55954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/57deb16e7d3c0ee611d4d267662217e41cb55954"}]},"branch":"refs/heads/master"},"90743c641fc28103ef69f9724a4529bbef627ac2":{"kind":"REWORK","_number":6,"created":"2020-08-30 16:16:53.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/6"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 16:16:48.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a TaskMonitor object so that the update job can be monitored for\ncompleteness. New TaskMonitor and Task objects are added according to\nthe Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/90743c641fc28103ef69f9724a4529bbef627ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/90743c641fc28103ef69f9724a4529bbef627ac2"}]},"branch":"refs/heads/master"},"561c4e622433557df2416c8ae112d267aedabc43":{"kind":"REWORK","_number":7,"created":"2020-08-30 16:20:00.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/7"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 16:19:48.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a TaskMonitor object so that the update job can be monitored for\ncompleteness. New TaskMonitor and Task objects are added according to\nthe Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/561c4e622433557df2416c8ae112d267aedabc43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/561c4e622433557df2416c8ae112d267aedabc43"}]},"branch":"refs/heads/master"},"55f889114c5a368f693a1de7fe105b2e59959c50":{"kind":"REWORK","_number":8,"created":"2020-08-30 19:32:44.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/8"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 19:30:05.000000000","tz":-240},"subject":"WIP: Fix firmware update to return a Task","message":"WIP: Fix firmware update to return a Task\n\nThis patch fixes the UpdateService.simple_update method so that it\nreturns a TaskMonitor object so that the update job can be monitored for\ncompleteness. New TaskMonitor and Task objects are added according to\nthe Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/55f889114c5a368f693a1de7fe105b2e59959c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/55f889114c5a368f693a1de7fe105b2e59959c50"}]},"branch":"refs/heads/master"},"826230841ac7d90f8d6258c8e61e28c4f62d30ea":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-09-01 13:57:19.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/9"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-01 13:56:45.000000000","tz":-240},"subject":"WIP: Allow monitoring progress of a firmware update","message":"WIP: Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/826230841ac7d90f8d6258c8e61e28c4f62d30ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/826230841ac7d90f8d6258c8e61e28c4f62d30ea"}]},"branch":"refs/heads/master"},"986ee372a3ddb6f44e1d8069533b110871a42558":{"kind":"REWORK","_number":10,"created":"2020-09-01 22:20:10.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/10"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-01 22:20:09.000000000","tz":-240},"subject":"WIP: Allow monitoring progress of a firmware update","message":"WIP: Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/986ee372a3ddb6f44e1d8069533b110871a42558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/986ee372a3ddb6f44e1d8069533b110871a42558"}]},"branch":"refs/heads/master"},"09676b8a526e81d7b73a1dad7b47d83411751f23":{"kind":"REWORK","_number":11,"created":"2020-09-02 01:22:57.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/11"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-02 01:22:47.000000000","tz":-240},"subject":"WIP: Allow monitoring progress of a firmware update","message":"WIP: Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/09676b8a526e81d7b73a1dad7b47d83411751f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/09676b8a526e81d7b73a1dad7b47d83411751f23"}]},"branch":"refs/heads/master"},"cd09a0c18cf71120e5231a010fa0f8e80349c9b3":{"kind":"REWORK","_number":12,"created":"2020-09-02 18:12:20.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/12"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-02 18:12:13.000000000","tz":-240},"subject":"WIP: Allow monitoring progress of a firmware update","message":"WIP: Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd09a0c18cf71120e5231a010fa0f8e80349c9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd09a0c18cf71120e5231a010fa0f8e80349c9b3"}]},"branch":"refs/heads/master"},"2216785c2a2e3f541b538f384ada339ec045ba07":{"kind":"REWORK","_number":13,"created":"2020-09-03 02:04:33.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/13"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-03 02:04:30.000000000","tz":-240},"subject":"WIP: Allow monitoring progress of a firmware update","message":"WIP: Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2216785c2a2e3f541b538f384ada339ec045ba07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2216785c2a2e3f541b538f384ada339ec045ba07"}]},"branch":"refs/heads/master"},"22f51f8d97dd7dea1b25a0de11c647dff0ba7131":{"kind":"REWORK","_number":14,"created":"2020-09-03 18:51:35.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/50/745950/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/14"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2020-09-03 18:51:34.000000000","tz":-240},"subject":"WIP: Allow monitoring progress of a firmware update","message":"WIP: Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nCo-Authored-By: Aija Jaunteva \u003caija.jaunteva@dell.com\u003e\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/22f51f8d97dd7dea1b25a0de11c647dff0ba7131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/22f51f8d97dd7dea1b25a0de11c647dff0ba7131"}]},"branch":"refs/heads/master"},"15c26b23a398d741946b9b1c5de44a0a042f87b4":{"kind":"REWORK","_number":15,"created":"2020-09-04 04:03:59.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/15"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-04 04:04:04.000000000","tz":-240},"subject":"Allow monitoring progress of a firmware update","message":"Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nCo-Authored-By: Aija Jaunteva \u003caija.jaunteva@dell.com\u003e\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/15c26b23a398d741946b9b1c5de44a0a042f87b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/15c26b23a398d741946b9b1c5de44a0a042f87b4"}]},"branch":"refs/heads/master"},"8273c0e4c5b4690dbc1dfa630c67092dc6bd3431":{"kind":"REWORK","_number":16,"created":"2020-09-04 15:16:31.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/16"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-04 15:15:49.000000000","tz":-240},"subject":"Allow monitoring progress of a firmware update","message":"Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nCo-Authored-By: Aija Jaunteva \u003caija.jaunteva@dell.com\u003e\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8273c0e4c5b4690dbc1dfa630c67092dc6bd3431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8273c0e4c5b4690dbc1dfa630c67092dc6bd3431"}]},"branch":"refs/heads/master"},"efd85b553bf41700091556f199665e06d4dfede0":{"kind":"REWORK","_number":17,"created":"2020-09-04 15:20:13.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/17"}}},"commit":{"parents":[{"commit":"3d4555771623e35fbf64cb40a41135d9aa8ed5b4","subject":"Fix retrieving software \u0026 firmware inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d4555771623e35fbf64cb40a41135d9aa8ed5b4"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-04 15:20:12.000000000","tz":-240},"subject":"Allow monitoring progress of a firmware update","message":"Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nCo-Authored-By: Aija Jaunteva \u003caija.jaunteva@dell.com\u003e\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/efd85b553bf41700091556f199665e06d4dfede0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/efd85b553bf41700091556f199665e06d4dfede0"}]},"branch":"refs/heads/master"},"0ed296162b05de17f4698fbd5d53c0c192228d89":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-09-04 19:33:11.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/18"}}},"commit":{"parents":[{"commit":"46b5d38d8417d64e45ed0b949647da10e64f695f","subject":"Add BIOS update apply time and maintenance window","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/46b5d38d8417d64e45ed0b949647da10e64f695f"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-04 19:33:17.000000000","tz":-240},"subject":"Allow monitoring progress of a firmware update","message":"Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nCo-Authored-By: Aija Jaunteva \u003caija.jaunteva@dell.com\u003e\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0ed296162b05de17f4698fbd5d53c0c192228d89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0ed296162b05de17f4698fbd5d53c0c192228d89"}]},"branch":"refs/heads/master"},"3f052a34982628d5dd2ffe5c4da742e5aa65dc75":{"kind":"REWORK","_number":19,"created":"2020-09-04 20:04:07.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/50/745950/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/50/745950/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/50/745950/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/50/745950/19"}}},"commit":{"parents":[{"commit":"46b5d38d8417d64e45ed0b949647da10e64f695f","subject":"Add BIOS update apply time and maintenance window","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/46b5d38d8417d64e45ed0b949647da10e64f695f"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-12 18:53:47.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-04 20:04:07.000000000","tz":-240},"subject":"Allow monitoring progress of a firmware update","message":"Allow monitoring progress of a firmware update\n\nThis patch updates the UpdateService.simple_update method so that it\nreturns a TaskMonitor object.  This allows the firmware update to be\nmonitored for completion. New TaskMonitor and Task objects are added\naccording to the Redfish Spec.\n\nCo-Authored-By: Aija Jaunteva \u003caija.jaunteva@dell.com\u003e\n\nChange-Id: I485d56a9804af723ddb55f8bc26f28a5ebefccc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3f052a34982628d5dd2ffe5c4da742e5aa65dc75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3f052a34982628d5dd2ffe5c4da742e5aa65dc75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
