)]}'
{"id":"openstack%2Fsushy~746963","triplet_id":"openstack%2Fsushy~master~I7b989bb106492a3158e0d7929871f0a7aa2b50bf","project":"openstack/sushy","branch":"master","topic":"test","hashtags":[],"change_id":"I7b989bb106492a3158e0d7929871f0a7aa2b50bf","subject":"DNM try breaking vmedia","status":"ABANDONED","created":"2020-08-19 15:36:13.000000000","updated":"2020-08-19 16:51:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f85240304db93c8c7884d37fc42af3a8c044baa","_number":746963,"virtual_id_number":746963,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-19 16:45:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 16:45:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e04f648e2372d737af2c3d69027ed80e5c64e482","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-19 15:36:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42875f314223efbf61c9c141b1d5da0372b49c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 16:45:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f2130df05d44f499f44ed4ef40d8083 : FAILURE in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c622fc89a34b450b84d1ccfdce14816d : FAILURE in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff841ef5b514f029767d62554df46b2 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d02d637c1afb47cebe91ce0949d5b558 : FAILURE in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7068c650be004a3fbcaf2c7591557f12 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255ba51c2d5148249bf99bfef8cc0c1c : SUCCESS in 6m 11s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/34909fe4ad0d4ac1b2d11f8017437c3d : SUCCESS in 49m 49s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ca3c8b9e238845999fd7054982e51cdf : FAILURE in 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"232587c27e1566980c278597c87bcc1611f68c65","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-19 16:51:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5ad232ec05f3161ad8eccf3a1dc1234c6e6775b","revisions":{"a5ad232ec05f3161ad8eccf3a1dc1234c6e6775b":{"kind":"REWORK","_number":1,"created":"2020-08-19 15:36:13.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/63/746963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/63/746963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/746963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/746963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/63/746963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/63/746963/1"}}},"commit":{"parents":[{"commit":"0e27eba2f55b027a5ec1a648753926d1e874eb85","subject":"Add a CI job with UEFI+vmedia and clean up the job definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0e27eba2f55b027a5ec1a648753926d1e874eb85"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-19 15:36:04.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-19 15:36:04.000000000","tz":120},"subject":"DNM try breaking vmedia","message":"DNM try breaking vmedia\n\nChange-Id: I7b989bb106492a3158e0d7929871f0a7aa2b50bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a5ad232ec05f3161ad8eccf3a1dc1234c6e6775b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a5ad232ec05f3161ad8eccf3a1dc1234c6e6775b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
