)]}'
{"id":"openstack%2Fsushy~748149","triplet_id":"openstack%2Fsushy~stable%2Frocky~Id8c5533d884eaf23b802122f7550791c5012d830","project":"openstack/sushy","branch":"stable/rocky","hashtags":[],"change_id":"Id8c5533d884eaf23b802122f7550791c5012d830","subject":"Do not log passwords and auth tokens when using SessionService","status":"MERGED","created":"2020-08-26 09:49:39.000000000","updated":"2020-09-10 11:49:29.000000000","submitted":"2020-09-10 11:48:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748149-1599738483460-2f639b6f","meta_rev_id":"9b7b3174b373a996a508c4d743f8ec7620beced8","_number":748149,"virtual_id_number":748149,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-10 10:02:01.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-09-08 20:33:34.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 11:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2020-08-31 12:07:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":2,"date":"2020-08-31 08:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":1,"date":"2020-08-31 08:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 08:14:47.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-08-31 12:07:59.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-09-08 20:33:34.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-09-10 10:02:01.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-09-10 11:48:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"307c20417ece49c10eb04581043540b2e7219949","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-26 09:49:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b79397a9fe12446beee69c16344b5fc65b0437ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:39:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b30de252e2e47c68d25faf016be1283 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cb17061d55442f59fdc6c2ed031dc75 : SUCCESS in 3m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c775bd7d836a43ec8e76651f392fc389 : SUCCESS in 4m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c9abe196189f47d8b1bfa441dfc3c3e4 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbba2e3973294adcb17b707fd8217587 : SUCCESS in 4m 04s\n- sushy-tempest-dsvm-ironic-ipa-partition-redfish-src https://zuul.opendev.org/t/openstack/build/ce4e009f0013401d9dd0e2eae5e35a39 : FAILURE in 42m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db48c43054534b7abe3c1b62b9bea05e : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da9fdeb45e4742598583faeeeb527fd1 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a6127f8e6d6acbeb0999664ab24ae38816fc1615","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-26 11:36:50.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-26:04:49/748149/1/ : FAILURE in 1h 38m 39s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-26:04:49/748149/1/ : FAILURE in 1h 46m 59s","accounts_in_message":[],"_revision_number":1},{"id":"17a1d2a416d1263b627f32f249afdef4c18d9e5b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-31 08:14:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b7bd0bf92d58b22936f7c3bb4cc97d4551d3006e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-31 08:16:24.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-31:03:14/748149/1/ : FAILURE in 1m 25s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-31:03:14/748149/1/ : FAILURE in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ce02dcefd1b89658c2dcb4e897701ba65406b61e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 08:27:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e26d489ce5149078ea3ac01be9373b0 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc565c20206d441e846337eebe3f827e : POST_FAILURE in 3m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2618bb10b534e818eaea99e3dcfee5f : POST_FAILURE in 3m 25s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/340c2175125e4e4fa22ad20fb35df093 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e72aa02f6dd42ca8afc75e94375e13a : SUCCESS in 3m 23s\n- sushy-tempest-dsvm-ironic-ipa-partition-redfish-src https://zuul.opendev.org/t/openstack/build/544eb8cf7822434eb69e6597c7d8f3d4 : FAILURE in 6m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74db799d8f4546f9b35a2cd961a7be4c : POST_FAILURE in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6900b2a65e394c5dbe4b40609abafa16 : POST_FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"55c5a2fea40f7bcd76ce654f1a40c887cf055052","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-08-31 12:07:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d753d31353469d88b860e51d76c41b4ba8bf0888","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-09-08 15:50:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea78608762a35625422cbb9edb8b54dbee528a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 16:43:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06789248e12a4d289c4f3db592b14b9b : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d415a5d3c864f44a849aa139412440b : SUCCESS in 3m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b5cb1560d38c4ae6baa7e7d5d5e61537 : SUCCESS in 5m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce82e7f7f64b49f8930fa9ad67e50d68 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99465fe40df4462881540d2453bb9481 : SUCCESS in 4m 55s\n- sushy-tempest-dsvm-ironic-ipa-partition-redfish-src https://zuul.opendev.org/t/openstack/build/334f299243d2406586937a9a104edc72 : FAILURE in 41m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73c68bed81c8419ca69b05fe758a58c1 : SUCCESS in 4m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78d42b62b2eb4834839afc7eab1aad4f : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3e109fbf7fef0e20756b805853a644c1e8b6c402","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-08 20:33:34.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-08:10:50/748149/1/ : FAILURE in 56m 55s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-08:10:50/748149/1/ : FAILURE in 1h 15m 29s","accounts_in_message":[],"_revision_number":1},{"id":"91b5d3be5b43aa1347f8582cb4f331463a31a01c","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-09-10 10:02:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff1ad5590ca36df7f558fadda0b50175468e6d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:00:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8b9adc540574a3fbbcfa771c601bb8d : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fb687b641a04c9b878de809f24bb2d1 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b55db143d7554805928ff1a3b00a64e5 : SUCCESS in 5m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/39105a3c4f1e49f28ad44c0e9066b6c8 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a559a0d4a9324b78a1b5e6d99c0f0b24 : SUCCESS in 5m 17s\n- sushy-tempest-dsvm-ironic-ipa-partition-redfish-src https://zuul.opendev.org/t/openstack/build/727b87b485f04b0d9cc2b9cb88694e23 : SUCCESS in 51m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4da1b7c56bb8457099fee0cef8e0c1d1 : SUCCESS in 6m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5953e61df08492baec86c996313b077 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"18bcba337f5439d4696283ae6c795ab6c8b4f6f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:00:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a7c2f8c06013af715a4eabc434c7535130eb617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:48:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb7fa74dea447d1928c23a013b4e385 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e097303df6f2438dbe2831418a917859 : SUCCESS in 2m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/655f8cb4f539470cb1f0bd85f745b316 : SUCCESS in 4m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c7e7f49d8b3c4d30b80b35fdcbd1307e : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f12433a99dbc459ea37568f006293cb0 : SUCCESS in 4m 06s\n- sushy-tempest-dsvm-ironic-ipa-partition-redfish-src https://zuul.opendev.org/t/openstack/build/3cb8aacc66cd41a0ac1521f78dbc685e : SUCCESS in 45m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40d49fea1e834c4587d8a6fd0bffae74 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b87c387512334c0fa4582f7d04308267 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"854bfc24f3efc6160ac087e2d9e2d55cd28f1fed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:48:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b7b3174b373a996a508c4d743f8ec7620beced8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:49:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34627a2d532a4ec39b784c44189e1a16 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fdcb88e352972ddbcbeea1ebb205e1d592745c9","revisions":{"2fdcb88e352972ddbcbeea1ebb205e1d592745c9":{"kind":"REWORK","_number":1,"created":"2020-08-26 09:49:39.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/49/748149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/748149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/748149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/748149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/748149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/748149/1"}}},"commit":{"parents":[{"commit":"e534332c1acf749d962bfa2015b16d9464f8dfbb","subject":"Disable HTTP connection pooling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e534332c1acf749d962bfa2015b16d9464f8dfbb"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-25 13:39:18.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-08-26 09:49:20.000000000","tz":120},"subject":"Do not log passwords and auth tokens when using SessionService","message":"Do not log passwords and auth tokens when using SessionService\n\nConflicts:\n\tsushy/connector.py\n\tsushy/tests/unit/test_utils.py\n\tsushy/utils.py\n\nChange-Id: Id8c5533d884eaf23b802122f7550791c5012d830\n(cherry picked from commit 5ffe8ee3b5ae569e5a2241cac5f25fae1bddad3e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2fdcb88e352972ddbcbeea1ebb205e1d592745c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2fdcb88e352972ddbcbeea1ebb205e1d592745c9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
