)]}'
{"id":"openstack%2Fsushy~748927","triplet_id":"openstack%2Fsushy~master~I045f82167a415c92d6b67532222aef89ab1d68ef","project":"openstack/sushy","branch":"master","topic":"parse_msgs","hashtags":[],"change_id":"I045f82167a415c92d6b67532222aef89ab1d68ef","subject":"Make message parsing more resilient","status":"MERGED","created":"2020-08-30 21:37:49.000000000","updated":"2020-09-24 01:23:08.000000000","submitted":"2020-09-23 19:46:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748927-1600890378141-b2ad87a5","meta_rev_id":"88ae0d3f734f2c13b3f7900052c0db481e676fc2","_number":748927,"virtual_id_number":748927,"owner":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"date":"2020-09-24 01:23:08.000000000","post_submit":true,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 19:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2020-09-23 00:23:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":2,"date":"2020-09-23 18:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":1,"date":"2020-09-23 18:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-30 21:39:45.000000000","updated_by":{"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},"reviewer":{"_account_id":30490,"name":"Digambar Patil","email":"digambar.patil@calsoftinc.com","username":"DigambarPatil"},"state":"REVIEWER"},{"updated":"2020-08-30 21:41:45.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-08-30 21:46:05.000000000","updated_by":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"reviewer":{"_account_id":26340,"name":"Ilya Etingof","email":"etingof@gmail.com","username":"etingof"},"state":"REVIEWER"},{"updated":"2020-08-31 08:17:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-09-03 13:42:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-09-03 14:24:13.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2020-09-03 16:15:25.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-09-04 21:28:03.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-09-09 00:16:05.000000000","updated_by":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"reviewer":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"state":"REVIEWER"},{"updated":"2020-09-22 01:17:52.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-09-23 00:23:09.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2020-09-23 18:40:14.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-09-23 19:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 01:23:08.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"}],"messages":[{"id":"2d7005e19ff81252b46b6eff733b83816630b035","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 21:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3c4a62a7996d0eb63d28a6ba179f4cd6bb2cf2e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-30 21:39:32.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-30:16:37/748927/1/ : FAILURE in 1m 31s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-30:16:37/748927/1/ : FAILURE in 1m 15s","accounts_in_message":[],"_revision_number":1},{"id":"16af5a4f049262d98383724159ae1394850cf78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 22:00:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05bbcbd963546dfa360ee99bedd5566 : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e4e715dd99f431b9725410e8d464659 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18aff555729a4e23b1a861fdab3431ad : SUCCESS in 2m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b29fb493e0b4edcb6387e79f1a42e33 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bce7c8cd592f4502bd550e511755aeeb : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5e059e4e7647fd9e263ed9ce7c42e7 : SUCCESS in 7m 37s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/748c01b8ac0d4745b64474747c1444a8 : FAILURE in 17m 10s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/47b6d02a50bf44c78f5568cb98106dbb : FAILURE in 21m 01s","accounts_in_message":[],"_revision_number":1},{"id":"765e47fde98367ec3ec2c3aabb37444d3939bad7","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-30 23:00:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"39e7682574b7ba6d0d0d16cd516a678993702d7a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-30 23:02:05.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-30:18:00/748927/1/ : FAILURE in 1m 30s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-30:18:00/748927/1/ : FAILURE in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"095753fee56dddc93267bd22af25d05063460b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 23:27:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7779ac01924e4d32a6daeefe71582c64 : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4bb6c073dee43929751cb4186b01c71 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c128b9d9a0441d08fa38b2e404a8a63 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ab57ad155c42b88ad2ec7e2e81c89c : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/778ceb86a86d4add8b810c0bd70109a7 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91ecf9721f14c0b8cceac4fc099f74d : SUCCESS in 5m 04s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/83bb4aa25398438bb6dd5a796b8d9038 : FAILURE in 20m 58s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/95cd5af85b834c448cf09e398b092921 : FAILURE in 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d18fc40e80ef8d910b3da7f58cb610177365478b","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-08-31 08:26:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nthank you for the patch. \n\nA question: is this allowed in Redfish spec or should the vendors fix this in their implementation in the end?","accounts_in_message":[],"_revision_number":1},{"id":"bc4646d814caa92866a30b78fb230d524542e0bc","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-31 12:59:15.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nAll of the issues that this patch addresses are problems with the vendor firmware where it is not complying with the Redfish spec. That said, I think it\u0027s preferable to do the best we can with message parsing instead of raising something like an index out of bound exception.","accounts_in_message":[],"_revision_number":1},{"id":"49a0f80455414c93e2bc28ec67da0f6f510c6d35","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-31 12:59:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"455f41b9e9ec9610219abfffd945f3484d1fba6f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-31 13:01:15.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-31:07:59/748927/2/ : FAILURE in 1m 26s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-31:07:59/748927/2/ : FAILURE in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5843780952a4c07c46e7621cee0b1ae9e77d128c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:30:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d1ec7eedd94089a075bbafc4ca6e4b : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2248c83b86174e73b1f03bdd8d32ffbc : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0566d7fe12441b9815fd3a7881e00f4 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99b5c24681334fceab17431676491a14 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c06e1c529f94eeb87d3cf67b17ca85b : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c76f1755693d4f72a26e678c2a59a67d : SUCCESS in 10m 24s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/eaa4414713804d1a8298cfd5de2cb23f : FAILURE in 19m 27s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/40b9ca7dec6f41f58ae5eb74bc013a09 : FAILURE in 22m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3230960285b69144beba57b9c00d030a75473008","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-08-31 13:53:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a5e57b045ff1a25b315bd29d1b74a9262498013c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-31 13:55:29.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-31:08:53/748927/2/ : FAILURE in 1m 28s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-31:08:53/748927/2/ : FAILURE in 1m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b9f74d1b28f34f0e8c8246ef6e400af084589891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 14:46:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/844dc813d2a9442d840dee4683eb671f : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b0ccb2e29f84807baddfe7f9a4f960c : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30ee456ff3b40ac90e189b3c5813491 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29583791ed424aee946863e3656bfd6d : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdbae5a07fd34c7ab0b9250957cc7b00 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8678a4059b004351babf54896edfc4e0 : SUCCESS in 8m 11s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e75e77ad720d44068b56dc9710d6da6c : FAILURE in 21m 45s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2745b9ee8a8e4fbeb3aa556e1c65124c : FAILURE in 21m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c666a29f6d55d2a9f475a4d84802b2c39c7ea8be","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-08-31 14:59:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5cc3415ee39288d9f48426198452617c76b5f081","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-09-01 20:53:47.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good.\nBesides improving resiliency does it cover any gaps for conformance to redfish spec?","accounts_in_message":[],"_revision_number":2},{"id":"fc65fa4b1f95eb5efd7a2d33639adf00de1b7840","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-02 12:10:47.000000000","message":"Patch Set 2:\n\nNo, the existing code complies with the redfish spec. This patch is to account for incorrect/incomplete implementation of the spec in BMCs.","accounts_in_message":[],"_revision_number":2},{"id":"bd41831c8375b6797a4d94a99cde25851f56d86e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 10:40:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nLGTM, but could you add a release note?","accounts_in_message":[],"_revision_number":2},{"id":"fd81295ff29a1596f050f2a9fc4b3fbb13ba0097","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-03 13:32:23.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nThanks for the review Dmitry!  Added a release note and addressed inline comments.","accounts_in_message":[],"_revision_number":2},{"id":"97f99f85156277d2e823ded75e6339ec94f5b438","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-03 13:32:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"04c180574cd80b91c0ad4acda6c4b329fb3988df","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 13:42:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d0d1355780888ce665a6cdf86f15af5981c53d75","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-09-03 14:24:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1fd7cdde9d0dfa47357b37da4f29319d949e49e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 14:39:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c7191caa394fcf9feb90a4a75904e1 : SUCCESS in 5m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d88bb1f21404b25957e4cce2273e17c : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95dab5a7bbe14546891036eabfb6ab34 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95a9ee87251843ec8108b753d0873d8b : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dde464a6dcb8424bb10c9642f9f2da69 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef5731266cf411c85a6dabc262331b5 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/464042fe8d7e4823a98e7076254e220e : SUCCESS in 7m 45s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/37de19378cc04aca814b417cc2bbd87b : SUCCESS in 48m 30s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/65538bc099be484e9da9fc7632bda121 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":3},{"id":"804761e7109a19ff99d58fff15ff11cb277d4641","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-03 15:02:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3fcf3c398ce5d5e1d3b659ba6c8e7cce85b129a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 15:02:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"751629eeb5d0ed67ca5878fc6aef582838a012c7","author":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"date":"2020-09-03 15:32:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nShould this pass the Dell Ironic CI before being merged?","accounts_in_message":[],"_revision_number":3},{"id":"e07379168151c2970dffda7dffaa39e169bf931c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-03 15:39:58.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"d18432055513a443fb3ceac84884b8b5e2b12e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:10:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46621f479ead46bda90aac1b9b7807c6 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c211b2c9a9049de8f070abaa15e154e : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f28e964b51bc40399ff7e55b65c75377 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/032f56b91dcb49838ae3e8bfa90f0a79 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5fd240e8b94f979e170f9547940b49 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e44ce7a9f641738a72ee1f18e9aede : SUCCESS in 6m 23s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/074d2ab3fb9144e094e8530258640c7b : SUCCESS in 56m 58s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7829d3bc5e554d1cb1dacd3715248a5e : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":3},{"id":"101749bb6847f32bf2a7447fd437c11c7221a476","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-03 16:15:20.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"56a5ca39c24991c28985808eee6536a4fbd38623","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-09-03 16:15:25.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"54cabf348f1c5b15e477c4347a1b23aa3774816d","author":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"date":"2020-09-03 17:50:33.000000000","message":"Patch Set 3: -Code-Review\n\n(2 comments)\n\nI have a couple more questions.","accounts_in_message":[],"_revision_number":3},{"id":"966ad3ac4b69ca8f64cfe608bc50b04d89af1116","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-03 17:54:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1372933ef416fa0d9d4a257c6f97599ef6397b68","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-03 18:02:38.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"139a73cd2b3e82c6b7c70487bcd7a75a09344ec9","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-04 05:03:43.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-03:08:32/748927/3/ : FAILURE in 1h 06m 44s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-03:08:32/748927/3/ : FAILURE in 1h 25m 30s","accounts_in_message":[],"_revision_number":3},{"id":"a41f436697d529d9fb8c0b4600566b04ef2ca85b","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-09-04 21:28:03.000000000","message":"Patch Set 3: Code-Review+1\n\nEric,\nplease, check Dell 3rd party CI failures.\n\nWonder if we need to check for the version of the redfish spec implementation if format is changing per redfish spec release.","accounts_in_message":[],"_revision_number":3},{"id":"4d389a7f706a90191730cdb7d06601e9f50dc963","author":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"date":"2020-09-09 00:16:05.000000000","message":"Patch Set 3:\n\ndell-recheck","accounts_in_message":[],"_revision_number":3},{"id":"41939373cd594af4c5755078f8786a2e21ba9818","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-09 01:50:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-08:19:15/748927/3/ : SUCCESS in 1h 11m 23s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-08:19:15/748927/3/ : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":3},{"id":"93c9660f11348480295f2e7020121f6d5fc14253","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-11 14:36:06.000000000","message":"Patch Set 3:\n\n\u003e Eric,\n \u003e please, check Dell 3rd party CI failures.\n \u003e \n \u003e Wonder if we need to check for the version of the redfish spec\n \u003e implementation if format is changing per redfish spec release.\n\nI advise extreme caution when trying to rely upon versions returned from the BMCs. Half of the implementations I\u0027ve seen the versioning has not matched actual fields presented nor aligns with actual behavior. Since it is such a moving target in many respects, it is somewhat to be expected but pure version reliance can likely have undesirable results. The approach we\u0027ve discussed in a larger sense is to try and see if $thing is there and try and handle that as we would expect and fallback where appropriate.","accounts_in_message":[],"_revision_number":3},{"id":"a422894bce455cef8634ace81c41f378a39d5d7f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-18 20:36:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71daaca0535482b350fd20f1d3b5564be96e4aac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 20:36:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b808d3bee17d7e4b82ffe0f2435db4f6d7047daa","author":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"date":"2020-09-18 21:04:04.000000000","message":"Patch Set 3: Code-Review+1\n\nThank you for answering my questions, Chris.\n\nSince Dell EMC Third-Party CI succeeded, LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"8d9686af6029a96d001a63fd98e7b1d4f910564b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 21:33:50.000000000","message":"Patch Set 3: Verified-2\n\n(7 comments)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10960322b2d24212a17d2aa1e20774c2 : FAILURE in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290a96e61eaf4edf974a0859ff5061ec : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b85abff30eca4a5fb93e8948f8c9bfd8 : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b09c58a624742f88db12ed006d69cfd : FAILURE in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8a6ad24fe24d2099033f212e33362e : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faf525bc06114774a9632768889318da : SUCCESS in 5m 06s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d562c470600c453a951d65952ea122ad : SUCCESS in 48m 05s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0ec26b252dd74b5188644f3b5ed5f460 : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f7cbf4e792da9b0f1df28bdda757fbab9e3482ac","author":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"date":"2020-09-20 23:11:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7c23d272f6c434c7786cf5728c88eae76453e063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-20 23:55:28.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b4a8acdfbca4844bf06c185d23df232 : FAILURE in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a34a141efc1e423a9f2d73fca4b60bb1 : FAILURE in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/703b614ce5ca496c8f58e025fb684e6a : FAILURE in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea23138774614b77a225d5e67552ea45 : FAILURE in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2461a6b33d7496894ad26a462bef772 : FAILURE in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5cc1c683cc4c4b996f6700d530a72c : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/830a76d1b6a7484bb7a36f73dbc1829c : SUCCESS in 3m 10s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4ed81fff563d45e7887fecc589d56164 : SUCCESS in 37m 00s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0d89f6da20934dbc84da0a93577dcc2f : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":3},{"id":"364aae390dad2e083fab88d6afe8f230c56d4cee","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-21 16:10:27.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-20:18:11/748927/3/ : FAILURE in 1m 54s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-20:18:11/748927/3/ : FAILURE in 1m 53s","accounts_in_message":[],"_revision_number":3},{"id":"fc0bfd607e2943c6821fe4d41877cc8d897a4236","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-22 17:48:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b664ea51b771e13ce854b1e5bbec8674a3dd7a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 18:51:42.000000000","message":"Patch Set 3:\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dcb4bb29f1b4ef9b01831de23529ec4 : FAILURE in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a8437d37dd4f639344aee45a0b07e4 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f2aaa5e6784b9ab8d88b6a9ab92e2c : FAILURE in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2b7498f75514fd9a23385b3b8b2fb11 : FAILURE in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf7911331a2f487ebb6f9cd66ae513c1 : FAILURE in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467192b8e09d47958c6f33ee62a27356 : SUCCESS in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecf8803ef9a343259eaf2b3b4813a5c3 : SUCCESS in 3m 46s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ea822af6694f4a40b593414c916b62d5 : SUCCESS in 46m 27s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/462f6d82978a4838b6bce820bfd762dc : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":3},{"id":"5a0ca14b746c4cb5583e3e9062f164f7c7403a40","author":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"date":"2020-09-22 22:33:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b292ca8eadf861b077243cc99508960bcef89136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:28:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee0950ed90e04ffd9b6304695a39782b : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b82b2b09e5e4ef2a9939fe014531029 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb0413f34cb45b88a0db3516a671d62 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19973fbeb3294766bff7894daf3e5238 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c14d24d6bdcc4941a4dd1945701e7b48 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f15d3222414dfca83853d7ad66873d : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d31d4860eeaf4426bc1e8a2330cae5af : SUCCESS in 3m 51s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/32fc33239e3c4cd9913d197312655cda : SUCCESS in 46m 33s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/31283f4eff984b3e91b2c6c5cc903307 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":4},{"id":"c359a613e2e078664e1db20ecc8d09bf729d049f","author":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"date":"2020-09-23 00:23:09.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks, again, Chris!","accounts_in_message":[],"_revision_number":4},{"id":"6e4a4c58e7678ebf2b5f8ff0b5d0d46818a66a63","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-23 18:40:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"dbc66d5e13ee165f89569b1fc21478f8180f3a1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 18:42:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5a149f20266edb3060de814edfb2e003533ab5ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:46:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0f86f53952f4a5892781a289e1e1c77 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9254fa026e04435964f212642be7caf : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e4e19ad25c54d779a84a19c9324d25d : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de99e2f41ae14e3998521ef5a3108727 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5082d49a8ad94cd3ad676c0cc2ab4d89 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e27dbe9337f54657a923ffbd2e1fa3a6 : SUCCESS in 3m 27s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e206f35f1e874ceea1d1517de93aed5a : SUCCESS in 51m 34s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/dde61be4d27a4135bba6aed516ddde68 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":4},{"id":"0a1817f2a5edc520e2b2d494fb68893dfe5cc22a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:46:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"725db9946917ce7b98c0e8851bc7d2f6d2fe9045","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 19:49:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e5bb1ebd7f4eee8eccc272afa508d6 : SUCCESS in 2m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe8450f29f9d476e871e221478ff2b52 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":4},{"id":"41b525595cececf31cab74b933eea76ebb11548b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-09-23 19:52:13.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit aab6812af4ba790d79a837c578aa761bced33d81","accounts_in_message":[],"_revision_number":4},{"id":"a720e61ce1fc6e11c52fac44ce2b27d060d18245","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-24 01:23:08.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-22:17:33/748927/4/ : SUCCESS in 58m 59s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-22:17:33/748927/4/ : FAILURE in 1h 47m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bad68e1e60ba2ea561e36394d1176e2315e3a907","revisions":{"1536c44a18629ca0d3141735ebab8b8dfcf2e1e3":{"kind":"REWORK","_number":1,"created":"2020-08-30 21:37:49.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/27/748927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/27/748927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/27/748927/1"}}},"commit":{"parents":[{"commit":"e730987a186be68de9555dfef66661f68b10c57a","subject":"Use Sessions URL from root service if it\u0027s provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e730987a186be68de9555dfef66661f68b10c57a"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 21:36:24.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 21:37:54.000000000","tz":-240},"subject":"Make message parsing more resilient","message":"Make message parsing more resilient\n\nFirmware from some vendors only includes the MessageKey in the\nMessageID.  In this case, fall back to finding the MessageKey first in\nthe Messages MessageRegistryFile and then in the BaseMessages\nMessageRegistryFile.\n\nIf the message can\u0027t be found, then set the parsed message to \u0027unknown\u0027\ninstead of throwing an exception.\n\nIf there are not enought arguments to populate the message, then\npopulate the missing arguments with \u0027unknown\u0027.\n\nChange-Id: I045f82167a415c92d6b67532222aef89ab1d68ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1536c44a18629ca0d3141735ebab8b8dfcf2e1e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1536c44a18629ca0d3141735ebab8b8dfcf2e1e3"}]},"branch":"refs/heads/master"},"123ab7a3c1f7764b9a17ff502da36669210b3d2b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-31 12:59:37.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/27/748927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/27/748927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/27/748927/2"}}},"commit":{"parents":[{"commit":"e730987a186be68de9555dfef66661f68b10c57a","subject":"Use Sessions URL from root service if it\u0027s provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e730987a186be68de9555dfef66661f68b10c57a"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 21:36:24.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-31 12:54:42.000000000","tz":-240},"subject":"Make message parsing more resilient","message":"Make message parsing more resilient\n\nFirmware from some vendors only includes the MessageKey in the\nMessageID. In this case, fall back to finding the MessageKey first in\nthe Messages MessageRegistryFile and then in the BaseMessages\nMessageRegistryFile.\n\nIf the message can\u0027t be found, then set the parsed message to \u0027unknown\u0027\ninstead of throwing an exception.\n\nIf there are not enough arguments to populate the message, then\npopulate the missing arguments with \u0027unknown\u0027.\n\nChange-Id: I045f82167a415c92d6b67532222aef89ab1d68ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/123ab7a3c1f7764b9a17ff502da36669210b3d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/123ab7a3c1f7764b9a17ff502da36669210b3d2b"}]},"branch":"refs/heads/master"},"7f92a862959ad166d3b2db811f2b5b65bb626c8d":{"kind":"REWORK","_number":3,"created":"2020-09-03 13:32:31.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/27/748927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/27/748927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/27/748927/3"}}},"commit":{"parents":[{"commit":"e730987a186be68de9555dfef66661f68b10c57a","subject":"Use Sessions URL from root service if it\u0027s provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e730987a186be68de9555dfef66661f68b10c57a"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 21:36:24.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-03 13:31:12.000000000","tz":-240},"subject":"Make message parsing more resilient","message":"Make message parsing more resilient\n\nFirmware from some vendors only includes the MessageKey in the\nMessageID. In this case, fall back to finding the MessageKey first in\nthe Messages MessageRegistryFile and then in the BaseMessages\nMessageRegistryFile.\n\nIf the message can\u0027t be found, then set the parsed message to \u0027unknown\u0027\ninstead of throwing an exception.\n\nIf there are not enough arguments to populate the message, then\npopulate the missing arguments with \u0027unknown\u0027.\n\nChange-Id: I045f82167a415c92d6b67532222aef89ab1d68ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7f92a862959ad166d3b2db811f2b5b65bb626c8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7f92a862959ad166d3b2db811f2b5b65bb626c8d"}]},"branch":"refs/heads/master"},"bad68e1e60ba2ea561e36394d1176e2315e3a907":{"kind":"REWORK","_number":4,"created":"2020-09-22 22:33:54.000000000","uploader":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"ref":"refs/changes/27/748927/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/27/748927/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/748927/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/27/748927/4"}}},"commit":{"parents":[{"commit":"b1bf4335120819268aa252762b98aede7b180958","subject":"Merge \"Make Actions field in Volume resource optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b1bf4335120819268aa252762b98aede7b180958"}]}],"author":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-08-30 21:36:24.000000000","tz":-240},"committer":{"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","date":"2020-09-22 22:33:55.000000000","tz":-240},"subject":"Make message parsing more resilient","message":"Make message parsing more resilient\n\nFirmware from some vendors only includes the MessageKey in the\nMessageID. In this case, fall back to finding the MessageKey first in\nthe Messages MessageRegistryFile and then in the BaseMessages\nMessageRegistryFile.\n\nIf the message can\u0027t be found, then set the parsed message to \u0027unknown\u0027\ninstead of throwing an exception.\n\nIf there are not enough arguments to populate the message, then\npopulate the missing arguments with \u0027unknown\u0027.\n\nChange-Id: I045f82167a415c92d6b67532222aef89ab1d68ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bad68e1e60ba2ea561e36394d1176e2315e3a907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bad68e1e60ba2ea561e36394d1176e2315e3a907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
