)]}'
{"id":"openstack%2Fsushy~748972","triplet_id":"openstack%2Fsushy~stable%2Fussuri~I825930623c84d3b6e5994e29f963b4f5af7afe4d","project":"openstack/sushy","branch":"stable/ussuri","topic":"no-guess-sessions-uri-stable/ussuri","hashtags":[],"change_id":"I825930623c84d3b6e5994e29f963b4f5af7afe4d","subject":"Use Sessions URL from root service if it\u0027s provided","status":"MERGED","created":"2020-08-31 08:16:48.000000000","updated":"2020-09-03 11:48:28.000000000","submitted":"2020-09-03 11:46:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748972-1599133573928-210ac784","meta_rev_id":"26dcf991957958e1ef4896e91e3f8db97a7ec8a2","_number":748972,"virtual_id_number":748972,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-02 15:47:48.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-09-03 02:52:43.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 11:46:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 10:35:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-03 10:35:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 02:52:43.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-09-03 10:35:30.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-09-03 11:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"433d8fc8ade76060e0c088fa8f2ae7c0bcfba256","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-31 08:16:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"63e469e03ec245a3647e9b56cd6ee925e9b2ec4e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-31 08:18:27.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-31:03:16/748972/1/ : FAILURE in 1m 07s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-31:03:16/748972/1/ : FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b718b42f143263e8fcf2f57c01a8ec9e31af6b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 08:34:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa88d0fc316545a19d9ed0d4152e310c : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e3102d38494b04bf746fd578fc899a : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643c3ff4ca354128bc735d68b9ab6ae4 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be3eb527bede4aaaa75a7eadf94f347b : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1732ada36c3442093204b8f8a6c249c : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969a1a77ed7643309f8fc1eada7475ee : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f51dd9fce2034aeda4a5787bace60df3 : SUCCESS in 4m 11s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/12aeb5556f37446eaf1a197bd76a6697 : FAILURE in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"78c867c3d2b9209685d69670f9b27672709be349","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-09-02 09:57:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f06bb66998e90bbdf194f3a64f2266a1c7f3a33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:46:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c9312ab720b4c3f81119930cc70acfc : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7782204f52d24be68131275c6dbc5b28 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704e1b0cea774f98960c5dd68daf9f40 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d1f739e690046a099002e1defd45d8b : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c6812008eae44c39c0761d65a67f6bd : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d651ca15400a483896aa0a5690978918 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c6709d2433043419dc0875d22222d98 : SUCCESS in 3m 55s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/bb9765fe0eb441d385d5eead0e5ac8e3 : TIMED_OUT in 1h 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"48f036dee23cacf62e19ad90d8585c62adaaf31b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-02 15:05:45.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-02:04:57/748972/1/ : SUCCESS in 52m 35s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-02:04:57/748972/1/ : FAILURE in 1h 47m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e65513929cbd53b3896879266ba319ced93bc0c5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-09-02 15:47:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"060163c15b0334f5185ef3ff2b774805f7b39a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:53:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/613b4f86dafe41a9a3b815d4a1c2c1d8 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5007b37a1a544d09b3b87d7b62c7eef : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/682b6db56a65483aa9ceae1ff6708132 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f0bd8e6cc54a78aabf951ae9bcf896 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c900966b5cf40dba03f3a38851ac3d6 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953692cb4b684662963cffc9819ea85c : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27746f9dd446436ba54f6338f13b622e : SUCCESS in 5m 03s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/6846a00f0c4841458838538a7fe6735e : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"30df39ecd19a3f131509a2eace031dfe25f967fb","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-03 02:52:43.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-02:10:47/748972/1/ : SUCCESS in 54m 38s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-02:10:47/748972/1/ : FAILURE in 1h 35m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f3645b858f2bb5c214a3b07457619dc25e547d30","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 10:35:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"855ddcb080ab4dfbe721fefa43fb5c234841a3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 10:35:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cff41e5a87b6e2c13efc1f8b72fdba0a92bb1cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:46:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21c076f021ae1b050c8aea3e893eb753519e7fe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:46:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6bc02f127034caeb8b951a66e4c66cd : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0656e14cbc64718b1fceef9309f11ba : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b5defdf5bb14cbd910b39edf63fcd8f : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e2cd73e5d884a2fb9cecfeaf61de00a : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08dd4eeef7e94056a5b168e3d762854c : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6753ce06e4c74b3f9e8ab9e33cb977cc : SUCCESS in 4m 03s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/8b8178ad2d9a43e58b9b17fb56d53376 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e4fee7465a52f2d964a74d5fd56ee7a19f94ee2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:48:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecfa18dd4f3841f6a480eed56821def0 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d0cb20382ea4126be4977618f81f689 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed883630ae6103ce1985efee9d9093d184ba4e7e","revisions":{"ed883630ae6103ce1985efee9d9093d184ba4e7e":{"kind":"REWORK","_number":1,"created":"2020-08-31 08:16:48.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/72/748972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/72/748972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/72/748972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/72/748972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/72/748972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/72/748972/1"}}},"commit":{"parents":[{"commit":"dc1b6132362edd7a06913e296f4e1922dde43192","subject":"Less scary warning when GET /redfish/v1/SessionService fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc1b6132362edd7a06913e296f4e1922dde43192"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-26 14:40:28.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-31 08:16:48.000000000","tz":0},"subject":"Use Sessions URL from root service if it\u0027s provided","message":"Use Sessions URL from root service if it\u0027s provided\n\nIf the root endpoint provides the link to Sessions, we should just\nuse it.\n\nChange-Id: I825930623c84d3b6e5994e29f963b4f5af7afe4d\nStory: 2008059\nTask: 40742\n(cherry picked from commit e730987a186be68de9555dfef66661f68b10c57a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ed883630ae6103ce1985efee9d9093d184ba4e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ed883630ae6103ce1985efee9d9093d184ba4e7e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
