)]}'
{"id":"openstack%2Fsushy~748973","triplet_id":"openstack%2Fsushy~stable%2Ftrain~I825930623c84d3b6e5994e29f963b4f5af7afe4d","project":"openstack/sushy","branch":"stable/train","topic":"no-guess-sessions-uri-stable/train","hashtags":[],"change_id":"I825930623c84d3b6e5994e29f963b4f5af7afe4d","subject":"Use Sessions URL from root service if it\u0027s provided","status":"MERGED","created":"2020-08-31 08:17:11.000000000","updated":"2020-09-03 11:51:59.000000000","submitted":"2020-09-03 11:49:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748973-1599133795695-1189cc01","meta_rev_id":"de3572d4dfcbbebe09c771b0f64723cf83933507","_number":748973,"virtual_id_number":748973,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-02 09:57:19.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2020-09-02 15:04:59.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 11:49:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-03 10:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-03 10:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 15:04:59.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-09-03 10:35:48.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-09-03 11:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"977de5e46636b54fe7cc8d8f1104fa535d437b9a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-08-31 08:17:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"68e722c01d342a00c4870d2a00c63482be04f25e","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-08-31 08:18:33.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-08-31:03:17/748973/1/ : FAILURE in 1m 10s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-08-31:03:17/748973/1/ : FAILURE in 1m 08s","accounts_in_message":[],"_revision_number":1},{"id":"253a34811722f0353f45c6778832b49ab27cecd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 09:23:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04eb61baf4ef4a34847c66e22a023c22 : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1f583eb9bd4f68b10a4233772f013b : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5434bbb8a3934fb196f80b13bdb8e390 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b97252b5896449e6a5d18f1d4b164441 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56ed07ae26644522945cd366ae6ae710 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/319d4a874d134c269550feee67136a7e : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc3e87ba91d4d71912d6c061db3ddc0 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30dc7f8aec3c488aa562babb4bac4e21 : SUCCESS in 3m 03s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/df47e502f4dd4512bee7b9843595f946 : FAILURE in 21m 47s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/f5c57e52511044d5859d5010ca4d115d : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0aefdc8c68c7fab0a58c7c63ec70aec3d470c42e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-09-02 09:57:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f74041c2089b5a38d5a51b75bd4d9fa31b645c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:10:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36837f9d370347b6b28760f522b7f46a : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc51df34370c427f8f3a0f37d8d45af7 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d92bdd96f844461aa6a6a4832064243 : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f530c80883641158edda2f22b62d167 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9fc6eb6ca024dd892778c9a17eb90dd : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b49c8f78dd64188b27d9d384196faa9 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10fbcfe7403e41c880130c5c283993e6 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1bd304e81c74e308f191e82933cb245 : SUCCESS in 5m 49s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/499fcbb749ac4c9cba2d6e0a1aba9b4c : SUCCESS in 1h 07m 21s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/20730a45932c4613950a90e537ac181a : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a2b568ec4b092f863ace0b719f508b3dcc8534eb","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-09-02 15:04:59.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-09-02:04:57/748973/1/ : FAILURE in 51m 20s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-09-02:04:57/748973/1/ : FAILURE in 1h 45m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e2b23e1cfdaab24ca270a535de45f209e99a4983","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-09-03 10:35:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0696d5324bb00c619007fa648af303fd1c34e127","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 10:36:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec465d4b64c98cfe2711ebab729d159acef46f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:49:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1862f07747e74a697d392cb444503a8dd4c87659","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:49:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/932ed2a054214fe3be36292a812c88bb : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d095048dfd59417ea02abd37a4146ed9 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c82fb613cd046dd88877ee4a7c66761 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/689ff063397d4548ae3a6d9553c5ec62 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fea3af33ac14c029edc587cfdcfbfb3 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba846b6b261444ca25216494f45243f : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4a5163d51f4ab3a34a4972af36c093 : SUCCESS in 3m 42s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/4ad69ea9fd304895a6a2e723ba88a8be : SUCCESS in 1h 05m 19s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/2e2fb8143fed4ede8769c7a0636434e0 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"595fa38a6a80cfc95bc948ae2d7f6d05cd6ada1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 11:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf2194766c0416db096d1b764d643ac : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ed53097f0eb4b5e833405c60745ac67 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e1507ac313c47832a6847a9e9874a317da884b0","revisions":{"3e1507ac313c47832a6847a9e9874a317da884b0":{"kind":"REWORK","_number":1,"created":"2020-08-31 08:17:11.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/73/748973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/73/748973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/748973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/748973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/73/748973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/73/748973/1"}}},"commit":{"parents":[{"commit":"226f932b5bb360585b7c7764f537d1a302a3216a","subject":"Less scary warning when GET /redfish/v1/SessionService fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/226f932b5bb360585b7c7764f537d1a302a3216a"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-26 14:40:28.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-08-31 08:17:11.000000000","tz":0},"subject":"Use Sessions URL from root service if it\u0027s provided","message":"Use Sessions URL from root service if it\u0027s provided\n\nIf the root endpoint provides the link to Sessions, we should just\nuse it.\n\nChange-Id: I825930623c84d3b6e5994e29f963b4f5af7afe4d\nStory: 2008059\nTask: 40742\n(cherry picked from commit e730987a186be68de9555dfef66661f68b10c57a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3e1507ac313c47832a6847a9e9874a317da884b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3e1507ac313c47832a6847a9e9874a317da884b0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
