)]}'
{"id":"openstack%2Fsushy~756906","triplet_id":"openstack%2Fsushy~master~I3e266902da9346119a0ea1b94c49a71fb2da0f7c","project":"openstack/sushy","branch":"master","topic":"taskservice-introduction","hashtags":[],"change_id":"I3e266902da9346119a0ea1b94c49a71fb2da0f7c","subject":"Adds basic support for TaskService to retrieve task information.","status":"MERGED","created":"2020-10-08 17:43:29.000000000","updated":"2020-12-03 17:28:55.000000000","submitted":"2020-12-03 17:26:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756906-taskservice-introduction","meta_rev_id":"eb222221bef321933457428b55dbd00f3d061aab","_number":756906,"virtual_id_number":756906,"owner":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2020-10-22 12:21:06.000000000","_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 17:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-11-17 06:06:56.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},{"value":2,"date":"2020-12-03 16:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":1,"date":"2020-10-23 01:50:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},{"value":1,"date":"2020-12-03 16:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 05:23:13.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-10-19 11:08:16.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-10-22 13:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 01:50:59.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-11-17 06:06:56.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-12-03 16:26:54.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"daf6e655c9d41e5ba841efd5c8e8649c7263b011","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-08 17:43:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22157326c135a288fbe7890192b2c7533115ebd4","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-08 17:45:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4a4d14868b6c0d1a1e6a29b1ae4a582f926fd51","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-08 18:29:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6939870ba4eec40484b9a1fd2b16c47d1a928a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:38:07.000000000","message":"Patch Set 3: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/851a87e764c043baa81f25480b43708a : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3ab879f4fe0495889f705bb2d9520e2 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/896fca7185914ba8acc5654c332a47af : FAILURE in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f466d78344ec48f99e484c51d34be96f : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/178a754c960d42ff887ecd4fb18320ac : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5dc6c9977741ea959016ec41004a65 : SUCCESS in 6m 03s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/da59438b24c648a6ab39ce90d8e97e21 : SUCCESS in 48m 41s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/15bdad0bdb7d4f99bb9074d8bdfc57aa : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c8f773bea3a2c437fc15d65e69cc8d5ef3e5a418","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-08 20:02:29.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-08:13:29/756906/3/ : SUCCESS in 57m 48s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-08:13:29/756906/3/ : FAILURE in 1h 25m 17s","accounts_in_message":[],"_revision_number":3},{"id":"e0c0c86eedb3d7115e8677092890eb3b425c1296","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-09 04:19:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"164974eb7297c8e16f9fd233551ad72406dfeb47","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-09 04:21:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4cec7d604a84864aab18aaaa023800d620341dba","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-09 04:23:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7e3803c6edda124ec768d5e9af2f0a0c0a7dc7f7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-09 05:23:13.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-08:23:23/756906/6/ : SUCCESS in 56m 35s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-08:23:23/756906/6/ : FAILURE in 59m 32s","accounts_in_message":[],"_revision_number":6},{"id":"51ed154328be8c1f771d48fedcedc649dfbbe84b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 05:28:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b631c909c5ab411888c379c61fa86239 : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27120904dbc748458eb9cbd0d8f8a1b3 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66b8d693a2542d8bfab6431e8479dae : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87d5bcc0dbe2493db0753223180b6952 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daf91478ff974eb584bd5286c4037759 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc279e1dd74241af9b37821f2d84ecdb : SUCCESS in 5m 36s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7167116e13a24f57aa370d406b97ea3f : SUCCESS in 52m 12s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/dcf03664b72b47ad81fc2c5d3d6cac10 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":6},{"id":"3159bab8c31fc3ca65dce3bd1ab23aef7523e71c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-19 11:08:16.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nLooks great, could you please add a release note? https://docs.openstack.org/ironic/latest/contributor/faq.html#create-a-new-release-note","accounts_in_message":[],"_revision_number":6},{"id":"bc04c2dea389ba2fac995b6312cf4574c1dc5e5c","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-22 12:19:33.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8a0af76d543cbd47f42f93e88d277ea75e99d8de","author":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"date":"2020-10-22 12:21:06.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nGreat! I added release notes and checked on your concern.","accounts_in_message":[],"_revision_number":7},{"id":"0a9064ebacef9ac893771702abc95ff32fb84981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 13:48:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14c1134d6bee422cb8777f1a665a6c0e : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c91f2616b964f96acec522562efd9d9 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4838f488460046c4aac444f4d075638a : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef3c237b41ee4e2a812d293d704b7220 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1040b2dd7e0743efa36ab7cee0b85e1d : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b379db8e303b49ee8d456d95b0f26bba : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d929d34ff1349c49dfb94d3e46e99a9 : SUCCESS in 6m 01s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/18f420b9423c4d07a621c944bb1c6642 : SUCCESS in 55m 37s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c9e7306106f2494c83fb02bf59248f6f : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":7},{"id":"04215501ba8ea8418669a841ad45d81cb09178bb","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-10-23 01:50:59.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"6057675d8940efb798e1ae20256278d11a0c8774","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-11-17 06:06:56.000000000","message":"Patch Set 7: Code-Review+2\n\nThank you Manuel","accounts_in_message":[],"_revision_number":7},{"id":"9512968f5d8009cdc9f6208ef995e7c35e8c9906","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-03 16:26:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"65bed2aa008e0f1e0a7ab99ff444d1db1525cdb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:27:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d010c496f6628ce4ef107b6d19defddc72f04bf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:26:45.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cfc54e02d9a41fe921dd9cf462f8cfb : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c3a7ca200c46fdafab55ad80ee09b4 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bce84432b5a2411e8883532d8e16f749 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d333a6fbfb54cbd8729a74815c15569 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6d97282f08468fab0169043838afe8 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1d47e6af703446caeb4c86724ace378 : SUCCESS in 3m 51s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/db7497877f364a1ba783adc3c42306e4 : SUCCESS in 50m 24s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fff5750837ab47bbbcea2a09e46f91b6 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":7},{"id":"79f94a5a48498fc491cdce4430b6525fb627771d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:26:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"eb222221bef321933457428b55dbd00f3d061aab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:28:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8753fcf7d2dc4ccb8818e5ec1ffffd2f : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b732d2fe6496433ebc3869e8fec30c4c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9d9dae7d019c6a55a2c78281bd349f1ebc45d90a","revisions":{"a9a121b2c82539bab6e7427485efeb1f7c41cc86":{"kind":"REWORK","_number":1,"created":"2020-10-08 17:43:29.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/1"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 17:42:43.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a9a121b2c82539bab6e7427485efeb1f7c41cc86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a9a121b2c82539bab6e7427485efeb1f7c41cc86"}]},"branch":"refs/heads/master"},"777bcc35c00a0c72b523dc820043eb8199d41aa5":{"kind":"REWORK","_number":2,"created":"2020-10-08 17:45:11.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/2"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 17:44:51.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/777bcc35c00a0c72b523dc820043eb8199d41aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/777bcc35c00a0c72b523dc820043eb8199d41aa5"}]},"branch":"refs/heads/master"},"27ca3cce990dc056438201656f7b22b439fa5fe8":{"kind":"REWORK","_number":3,"created":"2020-10-08 18:29:40.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/3"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 18:29:25.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/27ca3cce990dc056438201656f7b22b439fa5fe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/27ca3cce990dc056438201656f7b22b439fa5fe8"}]},"branch":"refs/heads/master"},"6675517601bf9b47e87417abc92fd70b88b76c04":{"kind":"REWORK","_number":4,"created":"2020-10-09 04:19:17.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/4"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-09 04:19:04.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6675517601bf9b47e87417abc92fd70b88b76c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6675517601bf9b47e87417abc92fd70b88b76c04"}]},"branch":"refs/heads/master"},"3dae339b0e268e68e191e62b0e4a8d5a9a64e4ee":{"kind":"REWORK","_number":5,"created":"2020-10-09 04:21:46.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/5"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-09 04:21:34.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3dae339b0e268e68e191e62b0e4a8d5a9a64e4ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3dae339b0e268e68e191e62b0e4a8d5a9a64e4ee"}]},"branch":"refs/heads/master"},"5f1ae108cc4d0714e3dbcb7acff6df8a9d1c3ead":{"kind":"REWORK","_number":6,"created":"2020-10-09 04:23:29.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/6"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-09 04:23:16.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5f1ae108cc4d0714e3dbcb7acff6df8a9d1c3ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5f1ae108cc4d0714e3dbcb7acff6df8a9d1c3ead"}]},"branch":"refs/heads/master"},"9d9dae7d019c6a55a2c78281bd349f1ebc45d90a":{"kind":"REWORK","_number":7,"created":"2020-10-22 12:19:33.000000000","uploader":{"_account_id":32469,"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","username":"mschoenlaub"},"ref":"refs/changes/06/756906/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/06/756906/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/06/756906/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/06/756906/7"}}},"commit":{"parents":[{"commit":"74b8111aff522cdddbb62438ad604d74fdfe2831","subject":"Merge \"Log only fields set in redfish response, not entire json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/74b8111aff522cdddbb62438ad604d74fdfe2831"}]}],"author":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-08 16:48:30.000000000","tz":120},"committer":{"name":"Manuel Schönlaub","email":"manuel.schoenlaub@gmail.com","date":"2020-10-22 12:19:12.000000000","tz":120},"subject":"Adds basic support for TaskService to retrieve task information.","message":"Adds basic support for TaskService to retrieve task information.\n\nChange-Id: I3e266902da9346119a0ea1b94c49a71fb2da0f7c\nSigned-off-by: Manuel Schönlaub \u003cmanuel.schoenlaub@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9d9dae7d019c6a55a2c78281bd349f1ebc45d90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9d9dae7d019c6a55a2c78281bd349f1ebc45d90a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
